)]}'
{"id":"openstack%2Fproject-config~693945","triplet_id":"openstack%2Fproject-config~master~Ie0008222e4f197d6e94727637ce092a77f6ea81e","project":"openstack/project-config","branch":"master","topic":"tripleo-operator-ansible","hashtags":[],"change_id":"Ie0008222e4f197d6e94727637ce092a77f6ea81e","subject":"Add tripleo-operator-ansible","status":"MERGED","created":"2019-11-12 23:14:02.000000000","updated":"2019-12-13 16:54:19.000000000","submitted":"2019-12-13 16:54:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693945-1576256059364-70b46864","meta_rev_id":"0496d68b69b231ea82cf189ee1ad69fc704e97d0","_number":693945,"virtual_id_number":693945,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 16:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-12-13 16:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-12-11 16:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-12-11 16:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-12-13 16:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2019-11-13 20:55:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 16:38:49.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-12-11 16:39:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-12-13 16:42:49.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-13 16:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4f8bde513e42588b78ce2d0514ac0a7a1865865","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-12 23:14:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2766502563482051e0024a39b275b89663983d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 23:25:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/749a3710c8a44493b0c1bb40747cc783 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a495126f4e5c484685dbe3c4e64b7d0f : FAILURE in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ab4fe5f38096ce4d3ef9150eaf7c2ff8113b0f91","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-12 23:28:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5bacb5066f1c7e61f1fd3494b5bebb5f0c8d684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 23:36:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4105c1a45b9e470fa448375a33d1dac5 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/58ed7fcb9966495db3617806069867cd : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4f1682c760fa40bcab1affa36067edbffc1ed2e2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-13 17:37:58.000000000","message":"Patch Set 2: Workflow-1\n\nlet\u0027s hold off on this until the spec discussion concludes","accounts_in_message":[],"_revision_number":2},{"id":"41532280c93adf3d912aa466946427da8fb94bd0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-13 20:54:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff0b4585f8367beadea3915e8929806400cb280b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-13 20:54:59.000000000","message":"Topic changed from tripleo-operators-ansible to tripleo-operator-ansible","accounts_in_message":[],"_revision_number":3},{"id":"a39d331abe0e5dbbcf03fac3019633a05a22ea85","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-13 20:55:51.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"f6fae94a74bb18428484fb17621c0593b6065537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 21:44:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6ccfda703a5415587990a0f2c2d4f72 : SUCCESS in 7m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a1a6e715889f47979616537e996228e7 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"390ad13196981fd75bc710639abeb324434e6ad4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-06 18:08:05.000000000","message":"Removed Workflow-1 by Alex Schultz \u003caschultz@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"474e9ac2481d52be96860d0047e43547b7c4141f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-10 13:33:08.000000000","message":"Patch Set 3:\n\nLGTM, waiting for PTL+1","accounts_in_message":[],"_revision_number":3},{"id":"cd06b3b27ff94aefe88f146c05f78b8471a35323","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-11 16:38:49.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5f586cad440657f74cbd1e216a9994a45663fdd0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-11 16:39:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f8ae205e7b97fc321d59a569c4d5ad2f2dacdb6b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-12-13 16:42:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"07859d5becf5df06ef0fe2d116e1a72ff41507e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:43:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fc6ab5b9947b98abb45579eff318198039ae395a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:54:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0496d68b69b231ea82cf189ee1ad69fc704e97d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:54:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/58ed981852d840c38b4396c189128cbb : SUCCESS in 8m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fe0fa0e35a034a8fb87fa4172abc72a2 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"423185e4581c1b4b2ec60ae7c44096f8d660ffe2","revisions":{"5eae72b6764c7363fd241e9aa74c934a70cb477d":{"kind":"REWORK","_number":1,"created":"2019-11-12 23:14:02.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/45/693945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/693945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/693945/1"}}},"commit":{"parents":[{"commit":"eb699621606db5f4a0f8ea060781a289af1cf4b5","subject":"Merge \"Remove python27 jobs from neutron dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb699621606db5f4a0f8ea060781a289af1cf4b5"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-12 23:11:37.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-12 23:13:23.000000000","tz":-420},"subject":"Add tripleo-operators-ansible","message":"Add tripleo-operators-ansible\n\nRelated-Blueprint: tripleo-operators-ansible\nNeeded-By: I67774ace48682da07fef5e1c8ba5c7f72181543c\nChange-Id: Ie0008222e4f197d6e94727637ce092a77f6ea81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5eae72b6764c7363fd241e9aa74c934a70cb477d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5eae72b6764c7363fd241e9aa74c934a70cb477d"}]},"branch":"refs/heads/master"},"a71e0a0fce7b9ae9ebfad1788e578c8dd52f1ab8":{"kind":"REWORK","_number":2,"created":"2019-11-12 23:28:45.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/45/693945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/693945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/693945/2"}}},"commit":{"parents":[{"commit":"eb699621606db5f4a0f8ea060781a289af1cf4b5","subject":"Merge \"Remove python27 jobs from neutron dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb699621606db5f4a0f8ea060781a289af1cf4b5"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-12 23:11:37.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-12 23:28:41.000000000","tz":-420},"subject":"Add tripleo-operators-ansible","message":"Add tripleo-operators-ansible\n\nRelated-Blueprint: tripleo-operators-ansible\nNeeded-By: I67774ace48682da07fef5e1c8ba5c7f72181543c\nChange-Id: Ie0008222e4f197d6e94727637ce092a77f6ea81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a71e0a0fce7b9ae9ebfad1788e578c8dd52f1ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a71e0a0fce7b9ae9ebfad1788e578c8dd52f1ab8"}]},"branch":"refs/heads/master"},"423185e4581c1b4b2ec60ae7c44096f8d660ffe2":{"kind":"REWORK","_number":3,"created":"2019-11-13 20:54:46.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/45/693945/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/693945/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/693945/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/693945/3"}}},"commit":{"parents":[{"commit":"eb699621606db5f4a0f8ea060781a289af1cf4b5","subject":"Merge \"Remove python27 jobs from neutron dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb699621606db5f4a0f8ea060781a289af1cf4b5"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-12 23:11:37.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-11-13 20:54:30.000000000","tz":-420},"subject":"Add tripleo-operator-ansible","message":"Add tripleo-operator-ansible\n\nRelated-Blueprint: tripleo-operator-ansible\nNeeded-By: I67774ace48682da07fef5e1c8ba5c7f72181543c\nChange-Id: Ie0008222e4f197d6e94727637ce092a77f6ea81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/423185e4581c1b4b2ec60ae7c44096f8d660ffe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/423185e4581c1b4b2ec60ae7c44096f8d660ffe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
