)]}'
{"id":"openstack%2Fproject-config~694478","triplet_id":"openstack%2Fproject-config~master~I0b4f091552cff894a130a89e9d55a902a25659a8","project":"openstack/project-config","branch":"master","topic":"neutron-interconnection-retire","hashtags":[],"change_id":"I0b4f091552cff894a130a89e9d55a902a25659a8","subject":"Retire neutron-interconnection project","status":"MERGED","created":"2019-11-15 08:45:05.000000000","updated":"2019-11-26 17:19:09.000000000","submitted":"2019-11-26 17:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694478-1574788749090-fe519b9b","meta_rev_id":"3c44c82c8f2626228688ad053fa87de5f8c9ca63","_number":694478,"virtual_id_number":694478,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 17:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-11-26 16:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-11-15 20:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2019-11-16 15:47:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-11-18 16:02:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-11-15 17:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-11-26 16:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 10:03:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-11-15 17:50:22.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-15 20:20:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-16 15:47:10.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-11-18 16:02:56.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-11-26 16:59:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-26 17:19:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce2fcfec7f9b50307037bb626a2c5e5db1839c32","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 08:45:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9816d192099df117ec95f83dba88f438697d5d59","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 08:53:32.000000000","message":"Patch Set 1:\n\nGovernance repo change https://review.opendev.org/#/c/694480/","accounts_in_message":[],"_revision_number":1},{"id":"eb3f7fab5826cc2c5781eb4880b4e3b75d25ef91","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 08:59:29.000000000","message":"Patch Set 1:\n\nLGTM - needs to wait for merge until we schedule the rename since renaming involves some manual steps.\n\nWill +2 once we\u0027re ready","accounts_in_message":[],"_revision_number":1},{"id":"958e559f1add55459a04b23d3f001842fe157aa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/db515229f4a14cd19277cfa607f54fe7 : SUCCESS in 6m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b535e10005824933bd0ac0391b51ebbb : SUCCESS in 6m 53s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d7f6330900164a4fbfa313cd5ca11e78 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dd06f35a88fad34c3511f2a5cd6a7378f1a0cf0e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-11-15 09:51:10.000000000","message":"Patch Set 1:\n\n(1 comment)\n\none question inline.","accounts_in_message":[],"_revision_number":1},{"id":"6703acb69133469f4322a026e2cd20af0e33ed1e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 09:53:22.000000000","message":"Patch Set 1:\n\nhttps://docs.openstack.org/infra/manual/creators.html#project-renames is the process for this - and this *renames* the repo.\n\nIt will keep it contents.","accounts_in_message":[],"_revision_number":1},{"id":"7110694e949c1a6498cbfafb00dfcd3a31e0bd80","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 09:54:48.000000000","message":"Patch Set 1: Code-Review-1\n\nas just discussed on IRC, we need to retire and recreate.","accounts_in_message":[],"_revision_number":1},{"id":"8eeb44ae4a4f91ceee31462cd25d0d85196a37e2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-11-15 10:03:26.000000000","message":"Patch Set 1:\n\nOr just retire it and not create a new project. If the project is dead, as the commit message suggests, then moving it to a different namespace so you can ignore it isn\u0027t really a good idea in the first place.","accounts_in_message":[],"_revision_number":1},{"id":"c1f8abc36a04d623a91dd9a938cb74944df1f630","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:48:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9f769e1ae0186374fc04e1c3d072a400289e26d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:48:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf1e9047e9a2b1489a13c6403deee67fd16d4633","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:48:22.000000000","message":"Patch Set 1:\n\nChanged to retire it simply","accounts_in_message":[],"_revision_number":1},{"id":"e18dc0333ae1f9d52d02125edcb1d006c0fff46b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:48:38.000000000","message":"Topic changed from project-rename to project-retire","accounts_in_message":[],"_revision_number":2},{"id":"ec12af7b021629a0341657286b0d3115b6ab8c0c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:49:54.000000000","message":"Topic changed from project-retire to neutron-interconnection-retire","accounts_in_message":[],"_revision_number":2},{"id":"f6fc8b043b15c3158fb5c81d208cdca23a01dada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 15:59:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b999e6bb39ec4f3b960e185d9a9e2f36 : SUCCESS in 8m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/361f7996368742de9635266f85948dde : SUCCESS in 3m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3e491d625a154e55b59f24e3d091ac2b : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c9f6d222c5500338d673987d3826fddfdff2f7a8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:04:02.000000000","message":"Patch Set 2: Code-Review-1\n\nSee:\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nbefore we can merge this, you need to remove all content from the repo and update the README. to do so, you first need to add noop-jobs here for this repo. so, this change is fine as *final* one (if it removes the noop-jobs again).","accounts_in_message":[],"_revision_number":2},{"id":"74ad28fcbd2aa6160bb706f42941234769aa0a83","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 16:07:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c61c711545a0f3dc1f8514f2985e7ce855e9fb29","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:09:56.000000000","message":"Patch Set 3:\n\nyou need to remove the noop-jobs again","accounts_in_message":[],"_revision_number":3},{"id":"8c35bc9cea2b8432f4a0928548ea245a2cd59caa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 16:11:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ee4f88c4c7021f5653c10c7390caf243c08eda6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:12:44.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"f1538f476c0b8991c6942e8b112966d927b93596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 16:22:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9bd8fa8b6fd4531ab8a01c8ae4bc009 : SUCCESS in 7m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eaea5e07a019407dbe9a43ad96577a8c : SUCCESS in 4m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7bfb1706d85d4839a99a87317bb7f3f0 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":4},{"id":"379e592a0cdebfecc5be7c5de8bf1201673bfa11","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:23:32.000000000","message":"Patch Set 4: -Code-Review\n\nbefore we can merge this, please abandon all open changes for the repo, see\nhttps://review.opendev.org/#/q/topic:neutron-interconnection-retire+is:open","accounts_in_message":[],"_revision_number":4},{"id":"00624d4e404c51fc463ebd994d70d5d7afed0a72","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:24:36.000000000","message":"Patch Set 4:\n\nhttps://review.opendev.org/#/q/project:openstack/neutron-interconnection+is:open is the link for open changes that needs to be an empty list before we can merge this change","accounts_in_message":[],"_revision_number":4},{"id":"7abf66657ca3c68ee04cd1e95450118f908e8b75","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-15 17:50:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ab59a09e79e4c5af769c60da9ea8a80872695abb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 20:20:56.000000000","message":"Patch Set 4: Code-Review+2\n\nall changes abandoned, this is good to go.","accounts_in_message":[],"_revision_number":4},{"id":"f45537dc338fbc1540e5f9dcaf46ce71f0b631cd","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-16 15:47:03.000000000","message":"Patch Set 4:\n\nAs said in other changes:\n\n\nI understand the need to adapt the project status to the lack of activity on the project in the past year.\nDuring this time, my time as a reviewer, preventing the code submitted to be merged, has been taken by other activites (still OpenStack related!).\n\nHaving been at the origin of the project, I have to apologize for the lack of communication of where we were about this project.\nMy apologies for that.\n\nWe still would like to have a place to let the proposal exit, code be reviewed and tested.\nHosting under \"x/\" would work for us.\n\nHope that this can work like this...\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"4e8029e40526f1adc284373c99b672e6e4cad4a5","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-16 15:47:10.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"6edcaf2113366a1bf92b61fdee55b9aba7a7ea32","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-16 16:29:07.000000000","message":"Patch Set 4:\n\nThomas, this change is according to the way we now move repos out of openstack namespace according to a recent TC policy.\n\nYou can create a *new* one to import the content into the \"x\" namespace if you want to continue to use it.","accounts_in_message":[],"_revision_number":4},{"id":"8d9b216725ddbd377451b79bf6ddeea5be20a7c9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-11-18 16:02:56.000000000","message":"Patch Set 4: Code-Review+1\n\nI agree with Andreas. According the new TC solution on project retirement, an existing repository under TC governance will drop all contents and if needed a new repository can be created under \"x\" namespace. No renaming happens. Thus, you can create a new repository under \"x\" namespace any time you want it. Specifying \"upstream\" to the existing neutron-interconnection repo allows you to restore all histories in the new \"x\" repository.","accounts_in_message":[],"_revision_number":4},{"id":"8bad9f43dd0ba5fae213c9dd94eab1a93188304d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-26 16:43:53.000000000","message":"Patch Set 4:\n\nhttps://review.opendev.org/#/c/696116/ was proposed, so I ignore the -1 by Thomas.","accounts_in_message":[],"_revision_number":4},{"id":"d40eeb1b4d9fd268490cc071f12203ac4ac8d82b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-26 16:59:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d348ac65a49fe536c5314475ebc41480ce639c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:59:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a01d5e5828f4d09c2837cb2f10b8a77ecb44a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 17:19:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e4bd9fc6217d47508e0516ad30e4ae7c : SUCCESS in 7m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/940b5fb0bdc240a9a0bc63c5459289a5 : SUCCESS in 4m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4f3af2358529497fa4f91b540a328db3 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4},{"id":"3c44c82c8f2626228688ad053fa87de5f8c9ca63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 17:19:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6197ea68785064a781db81a1c282de9bf2339331","revisions":{"c6fafee7df42b528742aac8501bca05e4901000b":{"kind":"REWORK","_number":1,"created":"2019-11-15 08:45:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/694478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/694478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/694478/1"}}},"commit":{"parents":[{"commit":"448a7848ea42e6d6469280fd3034d0d9c6b970bd","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/448a7848ea42e6d6469280fd3034d0d9c6b970bd"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:42:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:42:26.000000000","tz":60},"subject":"Rename neutron-interconnection from openstack/ to x/ namespace","message":"Rename neutron-interconnection from openstack/ to x/ namespace\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to move it out of stadium to x/ namespace.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\n[1] https://review.opendev.org/#/c/694466/\n\nChange-Id: I0b4f091552cff894a130a89e9d55a902a25659a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c6fafee7df42b528742aac8501bca05e4901000b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c6fafee7df42b528742aac8501bca05e4901000b"}]},"branch":"refs/heads/master"},"14edce42890f86da84ae020ceb5a29a6c47246fb":{"kind":"REWORK","_number":2,"created":"2019-11-15 15:48:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/694478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/694478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/694478/2"}}},"commit":{"parents":[{"commit":"448a7848ea42e6d6469280fd3034d0d9c6b970bd","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/448a7848ea42e6d6469280fd3034d0d9c6b970bd"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:42:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 15:47:20.000000000","tz":60},"subject":"Retire neutron-interconnection project","message":"Retire neutron-interconnection project\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to retire it.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\n[1] https://review.opendev.org/#/c/694466/\n\nChange-Id: I0b4f091552cff894a130a89e9d55a902a25659a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14edce42890f86da84ae020ceb5a29a6c47246fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14edce42890f86da84ae020ceb5a29a6c47246fb"}]},"branch":"refs/heads/master"},"3bc28a7fd2cf59a9ea45845e084c8cf193343054":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-15 16:07:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/694478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/694478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/694478/3"}}},"commit":{"parents":[{"commit":"cec4261dd02c4065ae910189ede3473beb2bf04b","subject":"Add noop-jobs template for neutron-interconnection project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cec4261dd02c4065ae910189ede3473beb2bf04b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:42:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 16:07:43.000000000","tz":0},"subject":"Retire neutron-interconnection project","message":"Retire neutron-interconnection project\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to retire it.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\n[1] https://review.opendev.org/#/c/694466/\n\nChange-Id: I0b4f091552cff894a130a89e9d55a902a25659a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bc28a7fd2cf59a9ea45845e084c8cf193343054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bc28a7fd2cf59a9ea45845e084c8cf193343054"}]},"branch":"refs/heads/master"},"6197ea68785064a781db81a1c282de9bf2339331":{"kind":"REWORK","_number":4,"created":"2019-11-15 16:11:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/694478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/694478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/694478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/694478/4"}}},"commit":{"parents":[{"commit":"cec4261dd02c4065ae910189ede3473beb2bf04b","subject":"Add noop-jobs template for neutron-interconnection project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cec4261dd02c4065ae910189ede3473beb2bf04b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:42:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 16:08:28.000000000","tz":60},"subject":"Retire neutron-interconnection project","message":"Retire neutron-interconnection project\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to retire it.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\n[1] https://review.opendev.org/#/c/694466/\n\nDepends-On: https://review.opendev.org/#/c/694490/\n\nChange-Id: I0b4f091552cff894a130a89e9d55a902a25659a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6197ea68785064a781db81a1c282de9bf2339331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6197ea68785064a781db81a1c282de9bf2339331"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
