)]}'
{"id":"openstack%2Fproject-config~694834","triplet_id":"openstack%2Fproject-config~master~I5ab80479ea362168ffeab99ff52c6430f975e0a3","project":"openstack/project-config","branch":"master","topic":"gerritBotAirship","hashtags":[],"change_id":"I5ab80479ea362168ffeab99ff52c6430f975e0a3","subject":"Add v2 branch for monitoring by IRC gerritbot","status":"MERGED","created":"2019-11-18 19:13:35.000000000","updated":"2019-11-19 20:30:13.000000000","submitted":"2019-11-19 20:30:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694834-1574195413037-64a7fc44","meta_rev_id":"b28245a12caa710a5f8e19135e18b48e48ae8c91","_number":694834,"virtual_id_number":694834,"owner":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-18 20:45:45.000000000","_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 20:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":2,"date":"2019-11-19 20:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-11-19 19:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":1,"date":"2019-11-19 20:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 19:15:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-19 20:21:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-19 20:30:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e181caaaa1d56ccfda6878c75c69c79d0649673","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2019-11-18 19:13:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48af188e3c66dc40afcbd43a759032dfc01577b5","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2019-11-18 19:14:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"288f52decbffe386e256b7414c1b689199c8392a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:27:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98322158d58440f9ac42d0816905846e : SUCCESS in 6m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/b315fd05388048f69e8d7e4898060114 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4ca590352a701518b9dbbda05374b773b9d6186d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-18 19:55:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf36dc820b1e0bf46e498192a228ae897dd17c9e","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2019-11-18 20:45:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfbe30092365e3508a019948ab3fdf8df050a7a7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-19 19:15:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41730bef7137e04334546527defc93f960847c4b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-19 20:21:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d454fc471aef253369d79cfb119958634b04cd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 20:21:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60bba1219301ce8dc84e96b7a12da6cacc0ce44c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 20:30:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0119692d654749bdb34baa5f79cbdbef : SUCCESS in 6m 29s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1ce4f5fbf7574c5dad482a16e9f540af : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"00336d9fd79d4836b3b8856e32e62e251f0a4354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 20:30:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95f10dd870e789f1feba2111da8a7cd2290275e5","revisions":{"ba5fa8486ab028a49241d2c3b32cae67edb09d9d":{"kind":"REWORK","_number":1,"created":"2019-11-18 19:13:35.000000000","uploader":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"ref":"refs/changes/34/694834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/694834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/694834/1"}}},"commit":{"parents":[{"commit":"cec4261dd02c4065ae910189ede3473beb2bf04b","subject":"Add noop-jobs template for neutron-interconnection project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cec4261dd02c4065ae910189ede3473beb2bf04b"}]}],"author":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2019-11-18 19:12:49.000000000","tz":60},"committer":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2019-11-18 19:12:49.000000000","tz":60},"subject":"Add v2 branch for monitoring by IRC gerritbot","message":"Add v2 branch for monitoring by IRC gerritbot\n\nChange-Id: I5ab80479ea362168ffeab99ff52c6430f975e0a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba5fa8486ab028a49241d2c3b32cae67edb09d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba5fa8486ab028a49241d2c3b32cae67edb09d9d"}]},"branch":"refs/heads/master"},"95f10dd870e789f1feba2111da8a7cd2290275e5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-18 19:14:37.000000000","uploader":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"ref":"refs/changes/34/694834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/694834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/694834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/694834/2"}}},"commit":{"parents":[{"commit":"cec4261dd02c4065ae910189ede3473beb2bf04b","subject":"Add noop-jobs template for neutron-interconnection project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cec4261dd02c4065ae910189ede3473beb2bf04b"}]}],"author":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2019-11-18 19:12:49.000000000","tz":60},"committer":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2019-11-18 19:13:46.000000000","tz":60},"subject":"Add v2 branch for monitoring by IRC gerritbot","message":"Add v2 branch for monitoring by IRC gerritbot\n\nThis v2 branch exists only in airship/treasuremap repository so far.\n\nChange-Id: I5ab80479ea362168ffeab99ff52c6430f975e0a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95f10dd870e789f1feba2111da8a7cd2290275e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95f10dd870e789f1feba2111da8a7cd2290275e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
