)]}'
{"id":"openstack%2Fproject-config~695843","triplet_id":"openstack%2Fproject-config~master~Idc6a38b555bebb2f69d6de4ef502ee58f010399b","project":"openstack/project-config","branch":"master","topic":"zuul-errors","hashtags":[],"change_id":"Idc6a38b555bebb2f69d6de4ef502ee58f010399b","subject":"Add missing repos for zuul tenant","status":"MERGED","created":"2019-11-25 07:26:26.000000000","updated":"2019-11-27 05:48:55.000000000","submitted":"2019-11-27 05:48:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695843-1574833735058-de65efff","meta_rev_id":"35a427e078c211983b90507877e6202880f2d533","_number":695843,"virtual_id_number":695843,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 05:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-11-27 05:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-27 02:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-11-27 05:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 17:27:50.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2019-11-27 00:27:45.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-27 02:31:57.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-11-27 05:48:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0257a2ca61b0a1d7b06aa37ab901508dbfe9f564","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-25 07:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f93a3927767f1ff6563749ff2d8fdcc4e6011a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 07:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e5d8a5fa5a1a45f99fea5d38c939c72a : SUCCESS in 9m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bf60f011f25d4a5298826e444ce9b8e8 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"08739d9d7aa1f4673163aa38f0ffe5db60b7ebed","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-11-26 17:27:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2f957cf8429263dbdbb1e90041da8f6d77bf9bd4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-11-26 21:48:14.000000000","message":"Patch Set 1: Code-Review+2\n\nI agree a bunch of these look like missing repos which this brings in; the list of errors matches this list.\n\nI don\u0027t understand some of the other errors, particularly\n\n---\n  Job dib-functests-base not defined\n\n  The error appears in the following job stanza:\n\n  job:\n      name: dib-functests-containers\n---\n\nit seems like that is all defined in the same file, i\u0027m not sure how it could be missing","accounts_in_message":[],"_revision_number":1},{"id":"59fd76dad019f9d98b6401101921a0f44b2b13a0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-27 00:27:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"307ad22dbd68b6742dd71f7069addaf7c3dc7433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 00:39:50.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"27961430a7db933f72a7ca536ae871369fff4261","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-11-27 02:31:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12c5d437786f65a89683dbfaed6fe1d9a270b1e8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-11-27 02:31:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f46eddf60660a26804f374e377128697f58aef28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:39:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/07f318dce6a342d1991954e1507a4967 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cd75909826124feea2b24bdbf9f3ae97 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"418e6c861a0ee45b299c0d36d593e65ed02415d4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-27 05:37:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nreapproving after rebase, taking previous +2 into account","accounts_in_message":[],"_revision_number":2},{"id":"3f3a334da8e59f6886ee02f85941109539ac13f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 05:38:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f3a5d7fdc6fe4f7861e6d3b9af1b526e4ce9fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 05:48:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f48f1ef083d471595387f100011e04a : SUCCESS in 5m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9ec4c4a6409a4c1294ec97004ed5a1f3 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0c90fbb0431f55c08c5b8ee6f4abd5015efbd856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 05:48:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a2e20b0985ab6296ca70650081e021b0d10eb55","revisions":{"c8c2ba98e60c615d2016424c925ee8569a8b8f98":{"kind":"REWORK","_number":1,"created":"2019-11-25 07:26:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/43/695843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/695843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/695843/1"}}},"commit":{"parents":[{"commit":"f57c760e427577c53ddd0773e3eda2ae12f2c4d1","subject":"Revert \"Revert \"Enable inap-mtl01\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f57c760e427577c53ddd0773e3eda2ae12f2c4d1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-11-25 07:25:16.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-11-25 07:25:16.000000000","tz":60},"subject":"Add missing repos for zuul tenant","message":"Add missing repos for zuul tenant\n\nWe have a couple of errors at\nhttps://zuul.opendev.org/t/zuul/config-errors due to missing\nrepositories, add them without config to allow using in\nrequired-projects.\n\nChange-Id: Idc6a38b555bebb2f69d6de4ef502ee58f010399b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8c2ba98e60c615d2016424c925ee8569a8b8f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8c2ba98e60c615d2016424c925ee8569a8b8f98"}]},"branch":"refs/heads/master"},"6a2e20b0985ab6296ca70650081e021b0d10eb55":{"kind":"REWORK","_number":2,"created":"2019-11-27 02:31:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/695843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/695843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/695843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/695843/2"}}},"commit":{"parents":[{"commit":"0be80ad3dbd107f8e12b0646a1d6547e132745dd","subject":"Merge \"zuul layout: include openstacksdk in zuul tenant for jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0be80ad3dbd107f8e12b0646a1d6547e132745dd"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-11-25 07:25:16.000000000","tz":60},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-11-27 02:31:14.000000000","tz":660},"subject":"Add missing repos for zuul tenant","message":"Add missing repos for zuul tenant\n\nWe have a couple of errors at\nhttps://zuul.opendev.org/t/zuul/config-errors due to missing\nrepositories, add them without config to allow using in\nrequired-projects.\n\nChange-Id: Idc6a38b555bebb2f69d6de4ef502ee58f010399b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a2e20b0985ab6296ca70650081e021b0d10eb55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a2e20b0985ab6296ca70650081e021b0d10eb55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
