)]}'
{"id":"openstack%2Fproject-config~696116","triplet_id":"openstack%2Fproject-config~master~I129d43eb599e02698451f7b18f08ad76f4cbb7a4","project":"openstack/project-config","branch":"master","topic":"neutron-interconnection-x-create","hashtags":[],"change_id":"I129d43eb599e02698451f7b18f08ad76f4cbb7a4","subject":"Create neutron-interconnection repo under x/ namespace","status":"MERGED","created":"2019-11-26 15:55:06.000000000","updated":"2019-11-27 15:20:27.000000000","submitted":"2019-11-27 15:11:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696116-1574867508193-ded4c5e8","meta_rev_id":"484cf69a4e039cc5662e6781920ed95d28ac31db","_number":696116,"virtual_id_number":696116,"owner":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 15:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-11-27 14:57:10.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-11-27 09:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-11-27 14:57:10.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 16:49:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-11-27 08:31:51.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-11-27 09:48:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-27 14:57:10.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-11-27 15:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b983e365105f23668145f06134eeb7df638bec5","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-11-26 15:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c56ccff3749118fa02edd4f754e4191da01b38e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:55:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"x/neutron-interconnection\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: x/neutron-interconnection\n      templates:\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 7071, column 3","accounts_in_message":[],"_revision_number":1},{"id":"02efeee3e761d6048502163fdf07a39e67a585fb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-11-26 16:49:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24ddd575f2fb70136cafec415e63dd4e2220204a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-26 17:08:00.000000000","message":"Patch Set 1: Code-Review-1\n\nWe first need to retire the repo with change https://review.opendev.org/694478 - please rebase this change once that change has merged.","accounts_in_message":[],"_revision_number":1},{"id":"82683a824fe3a2cb056e2f3fad8cc396f2a83213","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-27 08:31:51.000000000","message":"Patch Set 1:\n\n\u003e We first need to retire the repo with change https://review.opendev.org/694478\n \u003e - please rebase this change once that change has merged.\n\nHi Andreas,\n\nWill this still allow pulling the existing repo content in the new repo ?","accounts_in_message":[],"_revision_number":1},{"id":"f64cdfd707659b019daea899c8d9d757b5012819","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-27 09:18:23.000000000","message":"Patch Set 1:\n\nYou need to import the repo with the \"upstream\" keyword - follow the new project setup, see https://docs.openstack.org/infra/manual/creators.html","accounts_in_message":[],"_revision_number":1},{"id":"11cdb69b4cb541730b0cca2ba88ed191a2f1eeb2","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-11-27 09:22:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6a6dcb8e836b3999666c982d4a21017e5ae3bb3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-27 09:27:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6aa7d5982ff54977c656a8f48db59ecdf0dd3737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 09:36:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b75bcf263d334f4180078d2ed25c5096 : SUCCESS in 6m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0bd8bef750d943948708648345f638b3 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"70eb2820a131a4d4c50ac41fba8b9f40655dda50","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-11-27 09:41:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0163c0fe13b3c62346924f89102fcd977614b1f7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-27 09:48:16.000000000","message":"Patch Set 3: Code-Review+2\n\nThis should do it...","accounts_in_message":[],"_revision_number":3},{"id":"43471a9463d0d32ea5f5a54ae0a3a502326094df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 09:55:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/62bf4a531c7a48658172647f2bb85505 : SUCCESS in 8m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e74ccff2e70440d799bfb49a7d0aeb68 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5ba9c883523dc58a75837ec37289768adb48755a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-11-27 14:57:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3b741834160d26bf47f426892e2b7d1c841d3b52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:57:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ddb79ebcc30f637f72343406e5db136a5897241f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 15:11:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ed6de8e4d984d558a4faee77a6c8c3b : SUCCESS in 8m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e78d5e2dc765425a84b64679fc17eeb3 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"1087404b49bfa80fad4b534110278dfbd725ae98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 15:11:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a0cb58f6d69f02363e540f0647b679d89f5fe885","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-27 15:20:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fd5d5e2f3480f6ae38e205afbefda97632cb561d","revisions":{"e8718b1675cb7c2262e48ef0220b3bd04d17833f":{"kind":"REWORK","_number":1,"created":"2019-11-26 15:55:06.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/16/696116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/696116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/696116/1"}}},"commit":{"parents":[{"commit":"f57c760e427577c53ddd0773e3eda2ae12f2c4d1","subject":"Revert \"Revert \"Enable inap-mtl01\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f57c760e427577c53ddd0773e3eda2ae12f2c4d1"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-26 15:24:34.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-26 15:24:34.000000000","tz":60},"subject":"Create neutron-interconnection repo under x/ namespace","message":"Create neutron-interconnection repo under x/ namespace\n\nProject neutron-interconnection was removed from stadium, so create new repo\nunder x/ namespace to continue using it.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I129d43eb599e02698451f7b18f08ad76f4cbb7a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8718b1675cb7c2262e48ef0220b3bd04d17833f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8718b1675cb7c2262e48ef0220b3bd04d17833f"}]},"branch":"refs/heads/master"},"7018bdce00ea3e7df7d9747cd4654f64e021edf7":{"kind":"REWORK","_number":2,"created":"2019-11-27 09:22:03.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/16/696116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/696116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/696116/2"}}},"commit":{"parents":[{"commit":"ecd5828d84c6b22a5bc86b5851fcb663eadd811a","subject":"Merge \"Add missing repos for zuul tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecd5828d84c6b22a5bc86b5851fcb663eadd811a"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-26 15:24:34.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-27 09:24:47.000000000","tz":60},"subject":"Create neutron-interconnection repo under x/ namespace","message":"Create neutron-interconnection repo under x/ namespace\n\nProject neutron-interconnection was removed from stadium, so create new repo\nunder x/ namespace to continue using it.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I129d43eb599e02698451f7b18f08ad76f4cbb7a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7018bdce00ea3e7df7d9747cd4654f64e021edf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7018bdce00ea3e7df7d9747cd4654f64e021edf7"}]},"branch":"refs/heads/master"},"fd5d5e2f3480f6ae38e205afbefda97632cb561d":{"kind":"REWORK","_number":3,"created":"2019-11-27 09:41:55.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/16/696116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/696116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/696116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/696116/3"}}},"commit":{"parents":[{"commit":"ecd5828d84c6b22a5bc86b5851fcb663eadd811a","subject":"Merge \"Add missing repos for zuul tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ecd5828d84c6b22a5bc86b5851fcb663eadd811a"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-26 15:24:34.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-11-27 09:44:46.000000000","tz":60},"subject":"Create neutron-interconnection repo under x/ namespace","message":"Create neutron-interconnection repo under x/ namespace\n\nProject neutron-interconnection was removed from stadium, so create new repo\nunder x/ namespace to continue using it.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I129d43eb599e02698451f7b18f08ad76f4cbb7a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd5d5e2f3480f6ae38e205afbefda97632cb561d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd5d5e2f3480f6ae38e205afbefda97632cb561d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
