)]}'
{"id":"openstack%2Fproject-config~697576","triplet_id":"openstack%2Fproject-config~master~I5574ef2b92134871c0871302bc3f5a81edcf0357","project":"openstack/project-config","branch":"master","topic":"starlingx/oidc-auth-armada-app","hashtags":[],"change_id":"I5574ef2b92134871c0871302bc3f5a81edcf0357","subject":"starlingx: add oidc-auth-armada-app","status":"MERGED","created":"2019-12-05 22:16:31.000000000","updated":"2019-12-06 17:22:28.000000000","submitted":"2019-12-06 17:22:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697576-1575652948768-82587bb4","meta_rev_id":"90355e503a84b096303cc42df370bb0d6edbf1c6","_number":697576,"virtual_id_number":697576,"owner":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 17:22:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-06 05:48:10.000000000","_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-12-05 22:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-12-06 17:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-12-06 17:15:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 22:21:29.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-06 17:15:03.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-12-06 17:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fabe351bb2a0d88a79692b4c927d3377e81d1ce","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2019-12-05 22:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e92f4e03603cc3c0dc9ffe78d70b7ba30d38b4c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-12-05 22:21:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14e53f9d18ce722ae6206cd21856384f5e6cd495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:23:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bef5d3e9cd1f4670b2854d67ca4bc3e6 : SUCCESS in 5m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f1cfda5fb5ee4777b0551a0af84f5d5b : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c6dc4132fd0958edb71d9a609dbd5ea28c04a695","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-06 05:39:59.000000000","message":"Patch Set 1:\n\ndo you have any existing code you like to import with history - or want to start with an empty repository?","accounts_in_message":[],"_revision_number":1},{"id":"2378d6b59017625c6178d1b73a95f781bc94136b","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2019-12-06 05:48:10.000000000","message":"Patch Set 1:\n\n\u003e do you have any existing code you like to import with history - or\n \u003e want to start with an empty repository?\n\nThis is a new repo so empty to start with, thanks for checking!","accounts_in_message":[],"_revision_number":1},{"id":"60b92217df079e4bb011f2d0412134d6614d09c9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-06 17:15:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa693a63657d29bb297282be8f008edefb142ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:15:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7fa0f9bf1249a7a60a3b3da2e134978784d9f3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:22:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7fa86800d001e4016f630e292719fbef2b93b1e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:22:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e86ec8e1280542fbb2ad104b0a59cbeb : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/95227911a4d542508d15bdfbadd78b5c : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c790d2fb66075ec7c783d71d139f429fe17bc5a5","revisions":{"c790d2fb66075ec7c783d71d139f429fe17bc5a5":{"kind":"REWORK","_number":1,"created":"2019-12-05 22:16:31.000000000","uploader":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"ref":"refs/changes/76/697576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/697576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/697576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/697576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/697576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/697576/1"}}},"commit":{"parents":[{"commit":"14f812b1fb71fcf1cfbf1445cd03a18286ad3a9e","subject":"Merge \"Add `sushy-oem-idrac` project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/14f812b1fb71fcf1cfbf1445cd03a18286ad3a9e"}]}],"author":{"name":"Saul Wold","email":"sgw@linux.intel.com","date":"2019-12-05 22:12:18.000000000","tz":-480},"committer":{"name":"Saul Wold","email":"sgw@linux.intel.com","date":"2019-12-05 22:12:18.000000000","tz":-480},"subject":"starlingx: add oidc-auth-armada-app","message":"starlingx: add oidc-auth-armada-app\n\nThis repo will contain the helm charts and manifests for building\nthe OIDC Authentication app.\n\nChange-Id: I5574ef2b92134871c0871302bc3f5a81edcf0357\nSigned-off-by: Saul Wold \u003csgw@linux.intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c790d2fb66075ec7c783d71d139f429fe17bc5a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c790d2fb66075ec7c783d71d139f429fe17bc5a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
