)]}'
{"id":"openstack%2Fproject-config~698115","triplet_id":"openstack%2Fproject-config~master~I0c7542461555f5a23eddf0b3d3e548a734af778e","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I0c7542461555f5a23eddf0b3d3e548a734af778e","subject":"New project: go-redfish","status":"MERGED","created":"2019-12-09 23:59:56.000000000","updated":"2019-12-27 14:46:59.000000000","submitted":"2019-12-13 16:54:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698115-1576256057836-349857f7","meta_rev_id":"4f388a1a101594f2c40b2ca03b0279f88ad96e94","_number":698115,"virtual_id_number":698115,"owner":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 16:54:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-12-13 16:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-12-11 18:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-12-13 16:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-12-27 14:46:59.000000000","post_submit":true,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 18:49:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-12-13 16:41:58.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-13 16:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-27 14:46:59.000000000","updated_by":{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},"reviewer":{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},"state":"REVIEWER"}],"messages":[{"id":"c6ac7d50943d89541a7ae5b65b1175382081c865","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-12-09 23:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac05d2aa880a95df2817189f78ee6238a307087b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 00:08:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/95e09ec3dee94044b7f3a64298bae6be : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f880b5e28bf3411b9a4086239f9fa734 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d891041cd815dbda73358f931de44e8359f75eeb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-10 07:13:47.000000000","message":"Patch Set 1: Code-Review-1\n\nyou need to add to zuul/main.yaml","accounts_in_message":[],"_revision_number":1},{"id":"2b9044fe92d71efc497ae975e3c8e18bc45b2c7a","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-12-11 16:54:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec20678f6d334e9cdfaaaa82b047e5084f31e852","author":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"date":"2019-12-11 16:54:50.000000000","message":"Patch Set 1:\n\nAndreas: thanks, done","accounts_in_message":[],"_revision_number":1},{"id":"77f46393b3cdbc1868e884acb325625d91eedc28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:05:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9ffdd3c17c240c681a6e95cb4dda273 : SUCCESS in 6m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/42185aab1428430eb142d36309b2ce1a : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a62b458d818097eb8244ccadba6fb99d7899a048","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-11 18:49:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a76f181a15c135d7ab39d37f4f363fb0f49996e6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-12-13 16:41:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96432d9b36ffd229b0e048070f1f7e982f88bf07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:42:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48bbbf4e3edfe287f12c9d87a4e44b3e1672c2aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:54:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e6dd1dfdb51dda69671191a56dd6357120f320fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:54:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f71fede422574a7bacc0889ba027d782 : SUCCESS in 8m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d1a35db8787484cab349c724d612c0d : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04013d4943ac3329cc566373039ca9ff498e9275","revisions":{"5ef80ce3bd4d5e44514a7b6ed9b0dcf2d14f068f":{"kind":"REWORK","_number":1,"created":"2019-12-09 23:59:56.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/15/698115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/698115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/698115/1"}}},"commit":{"parents":[{"commit":"ac7ffa68080433742e9cd71d9e44e15bcbad4ba3","subject":"Merge \"starlingx: add oidc-auth-armada-app\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac7ffa68080433742e9cd71d9e44e15bcbad4ba3"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-12-09 23:55:53.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-12-09 23:55:53.000000000","tz":-360},"subject":"New project: go-redfish","message":"New project: go-redfish\n\nThis new project will host go client library for the Redfish\nhost management API.  It is intended to be consumable/useful\noutside of Airship itself; the Airship team created this library\nonly because of deficiencies in other available implementations.\n\nThe folks that own the github repo from which the extant\nimplementation should be imported have blessed this change, and\nwill continue their work under the Airship umbrella.\n\nPlease add me (Matt McEuen) as an initial member of the new ACL,\nand I can add additional seed core reviewers.\n\nThanks!\n\nChange-Id: I0c7542461555f5a23eddf0b3d3e548a734af778e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ef80ce3bd4d5e44514a7b6ed9b0dcf2d14f068f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ef80ce3bd4d5e44514a7b6ed9b0dcf2d14f068f"}]},"branch":"refs/heads/master"},"04013d4943ac3329cc566373039ca9ff498e9275":{"kind":"REWORK","_number":2,"created":"2019-12-11 16:54:42.000000000","uploader":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"ref":"refs/changes/15/698115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/698115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/698115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/698115/2"}}},"commit":{"parents":[{"commit":"ac7ffa68080433742e9cd71d9e44e15bcbad4ba3","subject":"Merge \"starlingx: add oidc-auth-armada-app\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac7ffa68080433742e9cd71d9e44e15bcbad4ba3"}]}],"author":{"name":"Matt McEuen","email":"matt.mceuen@att.com","date":"2019-12-09 23:55:53.000000000","tz":-360},"committer":{"name":"Matt McEuen","email":"madgin@madgin.net","date":"2019-12-11 16:54:24.000000000","tz":-360},"subject":"New project: go-redfish","message":"New project: go-redfish\n\nThis new project will host go client library for the Redfish\nhost management API.  It is intended to be consumable/useful\noutside of Airship itself; the Airship team created this library\nonly because of deficiencies in other available implementations.\n\nThe folks that own the github repo from which the extant\nimplementation should be imported have blessed this change, and\nwill continue their work under the Airship umbrella.\n\nPlease add me (Matt McEuen) as an initial member of the new ACL,\nand I can add additional seed core reviewers.\n\nThanks!\n\nChange-Id: I0c7542461555f5a23eddf0b3d3e548a734af778e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04013d4943ac3329cc566373039ca9ff498e9275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04013d4943ac3329cc566373039ca9ff498e9275"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
