)]}'
{"id":"openstack%2Fproject-config~699362","triplet_id":"openstack%2Fproject-config~master~I1fa7d5d38afb91f5bdd8bae0efe0ee8244cf22a8","project":"openstack/project-config","branch":"master","topic":"retire-fuel","hashtags":[],"change_id":"I1fa7d5d38afb91f5bdd8bae0efe0ee8244cf22a8","subject":"Retire fuel and fuel-ccp repositories (step 2)","status":"MERGED","created":"2019-12-17 07:54:50.000000000","updated":"2019-12-18 16:11:50.000000000","submitted":"2019-12-18 16:07:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699362-1576685254415-d94f54cc","meta_rev_id":"ae7eba42185bde697e74d9746e1af877ec61d0f7","_number":699362,"virtual_id_number":699362,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 16:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-12-18 14:46:46.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-12-18 15:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2019-12-18 09:30:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-12-18 15:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 08:37:15.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-12-17 18:49:20.000000000","updated_by":{"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},"reviewer":{"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},"state":"REVIEWER"},{"updated":"2019-12-18 08:42:57.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-12-18 09:30:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-18 14:46:46.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-12-18 15:53:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-18 16:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08c8050926ba8a9e50087c2d8615dc5148afafdd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-17 07:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8815b03f60163365ec445e00185c575872038322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 08:04:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f538731e0e2452b9199dc72ad4a6f51 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fa4ae89348a349bbb0fbd4280f9f5b701305acc4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-12-17 08:37:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b4620b79f1495d4c460d59ca75e56ba4b7684a7","author":{"_account_id":6786,"name":"Sergey Lukjanov","email":"me@slukjanov.name","username":"slukjanov"},"date":"2019-12-17 18:49:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c961bbaa5ea8f063d0ab852100fdfb0da4f3d9a4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 06:38:16.000000000","message":"Topic changed from fuel-cleanup-trusty to bye-trusty","accounts_in_message":[],"_revision_number":1},{"id":"ed397db5932129a75e745488eb2a067165787704","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 08:27:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d7894943bd51d8b98e48767c54915875464b7bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 08:40:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/fca6d42df0c3456ab48fc7561ebc9952 : SUCCESS in 4m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/939fd802173a4a819e5dc96fd7ff371f : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9d1160823ea4417b94a5d5c3c34cf4738efc6993","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-12-18 08:42:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"639c79f7c78d9a855853d80122e11fc1b5b88b3a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 09:21:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24072e3c907562dca1716d1fc53f94c261a3afc6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-18 09:30:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9d4739182e5fc9f7b1e88bc4e2afe993d34bf379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:37:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/47ee3deeb60d4996b8b75ab31565dc5a : SUCCESS in 4m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c654ca9580f24a59909c4f0d9da3646c : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":3},{"id":"d63118c30221c603fe7b485600c08f241c9f6d8c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 14:36:20.000000000","message":"Topic changed from bye-trusty to retire-fuel","accounts_in_message":[],"_revision_number":3},{"id":"b67055c69ecc09989452630ae9c20597a2516355","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-12-18 14:46:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a310562e834dbffcd9ad1fc989bbf971a07425d9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-12-18 15:53:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c31444444ecd71c212cd3d74675f70daa445355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 15:53:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0919202fb0195fd2754eea549fd15f2a0f34a3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:07:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/b1b5f1ae686f4862a3f25b7200d5ea3d : SUCCESS in 3m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0796da271aef4c7798c5a9a3543028d0 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8a66396200f9abd8722ce0433106ae6e43caacc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:07:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ae7eba42185bde697e74d9746e1af877ec61d0f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:11:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/2fa0a54cccc84b71a23167eb545b5c08 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"877777276d8f8b1a0c82c1f47970e36cdc7fe73d","revisions":{"f102dcfa45d59d7d1a0a59ca4b33064324afa7e5":{"kind":"REWORK","_number":1,"created":"2019-12-17 07:54:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/62/699362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/699362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/699362/1"}}},"commit":{"parents":[{"commit":"e8938e5e250e90e5f34e13d81fe9848999e073de","subject":"Update nodepool dib stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8938e5e250e90e5f34e13d81fe9848999e073de"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-17 07:51:29.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-17 07:51:29.000000000","tz":60},"subject":"Remove trusty fuel jobs","message":"Remove trusty fuel jobs\n\nRemove jobs from fuel repos that still use trusty nodes so that we can\ncleanup trusty from infrastructure.\nAdd noop-jobs template instead.\n\nNote that fuel was retired from OpenStack in 2017 and none of the\nrepositories touched here has any change merged since December 2017.\n\nChange-Id: I1fa7d5d38afb91f5bdd8bae0efe0ee8244cf22a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f102dcfa45d59d7d1a0a59ca4b33064324afa7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f102dcfa45d59d7d1a0a59ca4b33064324afa7e5"}]},"branch":"refs/heads/master"},"0c343fdbc9309229e8c6cf8c03fce5b138c17c23":{"kind":"REWORK","_number":2,"created":"2019-12-18 08:27:14.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/62/699362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/699362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/699362/2"}}},"commit":{"parents":[{"commit":"e8938e5e250e90e5f34e13d81fe9848999e073de","subject":"Update nodepool dib stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8938e5e250e90e5f34e13d81fe9848999e073de"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-17 07:51:29.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-18 08:24:26.000000000","tz":60},"subject":"Retire fuel and fuel-ccp repositories (step 2)","message":"Retire fuel and fuel-ccp repositories (step 2)\n\nRetire fuel from openstack namespace and fuel-ccp from x namespace,\nthese repositories are dead.\n\nAnnouncement:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-December/011647.html\n\nAlso remove fuel-specs from specs index page.\n\nThis is step 2 in\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nChange-Id: I1fa7d5d38afb91f5bdd8bae0efe0ee8244cf22a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c343fdbc9309229e8c6cf8c03fce5b138c17c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c343fdbc9309229e8c6cf8c03fce5b138c17c23"}]},"branch":"refs/heads/master"},"877777276d8f8b1a0c82c1f47970e36cdc7fe73d":{"kind":"REWORK","_number":3,"created":"2019-12-18 09:21:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/62/699362/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/699362/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/699362/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/699362/3"}}},"commit":{"parents":[{"commit":"e8938e5e250e90e5f34e13d81fe9848999e073de","subject":"Update nodepool dib stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8938e5e250e90e5f34e13d81fe9848999e073de"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-17 07:51:29.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-18 09:07:34.000000000","tz":60},"subject":"Retire fuel and fuel-ccp repositories (step 2)","message":"Retire fuel and fuel-ccp repositories (step 2)\n\nRetire fuel from openstack namespace and fuel-ccp from x namespace,\nthese repositories are dead.\nThis includes the tuning-box repo, it\u0027s part of fuel.\n\nAnnouncement:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-December/011647.html\n\nAlso remove fuel-specs from specs index page.\n\nThis is step 2 in\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nChange-Id: I1fa7d5d38afb91f5bdd8bae0efe0ee8244cf22a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/877777276d8f8b1a0c82c1f47970e36cdc7fe73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/877777276d8f8b1a0c82c1f47970e36cdc7fe73d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
