)]}'
{"id":"openstack%2Fproject-config~699664","triplet_id":"openstack%2Fproject-config~master~I76a6bce7efd2b0623f65810d7b481624c43cbd4e","project":"openstack/project-config","branch":"master","topic":"retire-fuel","hashtags":[],"change_id":"I76a6bce7efd2b0623f65810d7b481624c43cbd4e","subject":"Retire x/fuel-plugin-calico","status":"MERGED","created":"2019-12-18 11:15:40.000000000","updated":"2019-12-18 17:05:44.000000000","submitted":"2019-12-18 17:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699664-1576688744009-8ae24aaf","meta_rev_id":"f38b5caecf9381eb0d900cb6e3d762dbd9b5c6d3","_number":699664,"virtual_id_number":699664,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 17:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-12-18 16:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-12-18 16:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-12-18 16:52:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 16:28:11.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-12-18 16:52:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-12-18 17:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4aaf8078db8b3dfc6457385fb996764982acd371","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 11:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8466508c3ca8d3c10392acaea8038a46c3f23ff4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-18 11:17:22.000000000","message":"Topic changed from calico to retire-fuel","accounts_in_message":[],"_revision_number":1},{"id":"84ea18ab3e75302e227dd42012fc03a40cea8007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 11:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/daff7fe354544468974523b9ab1b8caa : SUCCESS in 7m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/450ef3bd67214bd1bcd401a5e18cf2d2 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cc754f9dbc6af46f83b65b3c5b03de2f3ccbab86","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-12-18 16:28:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f87c4adcb2653a5b76703cf39a1313add0e41e7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-12-18 16:52:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b92621361cbb40df0a4b9e8e21b12086abb20b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:53:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7df9a70727c3118548228c8a59ff0ae4cadaf6a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:05:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f77a099af37411899295a3895b37c4c : SUCCESS in 7m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3a84a0646d8645c5a1a8aaa9f8bcfd19 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f38b5caecf9381eb0d900cb6e3d762dbd9b5c6d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:05:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56793c6ccff3bf01ce9f0231cd4dcf783f242b7a","revisions":{"56793c6ccff3bf01ce9f0231cd4dcf783f242b7a":{"kind":"REWORK","_number":1,"created":"2019-12-18 11:15:40.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/64/699664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/699664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/699664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/699664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/699664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/699664/1"}}},"commit":{"parents":[{"commit":"8ebb00d291697c56809b90ef79139b34478498d5","subject":"Merge \"Revert \"Revert \"Disable fortnebula\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ebb00d291697c56809b90ef79139b34478498d5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-18 11:14:34.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-12-18 11:14:34.000000000","tz":60},"subject":"Retire x/fuel-plugin-calico","message":"Retire x/fuel-plugin-calico\n\nThis is finalizes retiring x/fuel-plugin-calico.\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2019-December/011653.html\n\nDepends-On: https://review.opendev.org/699662\nChange-Id: I76a6bce7efd2b0623f65810d7b481624c43cbd4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56793c6ccff3bf01ce9f0231cd4dcf783f242b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56793c6ccff3bf01ce9f0231cd4dcf783f242b7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
