)]}'
{"id":"openstack%2Fproject-config~701447","triplet_id":"openstack%2Fproject-config~master~Ic8320d4190eaacee28c7cdcd94cadf2b66313af9","project":"openstack/project-config","branch":"master","topic":"retire-dox","hashtags":[],"change_id":"Ic8320d4190eaacee28c7cdcd94cadf2b66313af9","subject":"Finish retiring x/dox (step 4)","status":"MERGED","created":"2020-01-07 19:56:23.000000000","updated":"2020-01-08 20:47:34.000000000","submitted":"2020-01-08 20:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701447-1578516454089-69f755c7","meta_rev_id":"7c707692cd684d6902e65d839520f73a49322104","_number":701447,"virtual_id_number":701447,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 20:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-08 20:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-08 20:11:34.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-08 20:12:27.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-01-08 20:38:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 20:07:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-08 20:12:27.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-08 20:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"242c1c1a968e9141827cb583f11779807dd241bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-07 19:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0c6f39fb70f78e2bc144689cec34d1212ca723a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-07 19:57:08.000000000","message":"Patch Set 1: Workflow-1\n\nwaiting for changes to get abandoned:\nhttps://review.opendev.org/#/q/project:x/dox+is:open","accounts_in_message":[],"_revision_number":1},{"id":"7d97e96ace328ba6795004cefcc397f73c81f1fc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-07 20:00:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f6b3cc0e2e0a4034559c371eed9a2157ee267f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 20:00:58.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"0f3aae3470787503eb1b9f9971873db30a1c147d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-08 20:07:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a16dbfbe41935a35a7c6da455fce7348be585445","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-08 20:11:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"71ae705b0665484a2de697f4e5c0224bbe53141d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 20:11:53.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3763b35e5e10152eef8beb24d4071e051259268f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 20:11:58.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"8dc996c8de14f92d66a8b5b7fb66788c5c51618b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-08 20:12:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e270d9de284270e2da34c00eb40b952fde7f771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:23:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ebf5849976e42088f7dfe6efbbb6831 : SUCCESS in 6m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1a5d5865e4e3475fa8f8f359b22888a2 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f6176d6d353226c5309bec73c809b1d091e72fbc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 20:38:50.000000000","message":"Patch Set 2: Workflow+1\n\ndependencies merged","accounts_in_message":[],"_revision_number":2},{"id":"aa0472a67f92bd4fd5f757ce9a1a8d264b79f8c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:39:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d04f36792c078ec6bd927d29d70a73ff7398120b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:47:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/edf90f8b5027499f91d8262eaeff7ec3 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/443fd8baf0c84464b68f4b880b018e65 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"749b23198bedb435640186614c0882483d7dc073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:47:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"185408332dae90a7f67e9781ed164aaf105a7880","revisions":{"62af65611b4aa3ccc6bbf9fb7ac1cd6407c8f6df":{"kind":"REWORK","_number":1,"created":"2020-01-07 19:56:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/47/701447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/701447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/701447/1"}}},"commit":{"parents":[{"commit":"21c5a7d6cb37e78f2c141157f4fd738b37b95b95","subject":"Retire x/dox (step 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21c5a7d6cb37e78f2c141157f4fd738b37b95b95"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-07 19:55:21.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-07 19:55:21.000000000","tz":60},"subject":"Finish retiring x/dox (step 4)","message":"Finish retiring x/dox (step 4)\n\nRemove x/dox from OpenDev and mark it as retired.\n\nDepends-On: https://review.opendev.org/701443\nChange-Id: Ic8320d4190eaacee28c7cdcd94cadf2b66313af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62af65611b4aa3ccc6bbf9fb7ac1cd6407c8f6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62af65611b4aa3ccc6bbf9fb7ac1cd6407c8f6df"}]},"branch":"refs/heads/master"},"185408332dae90a7f67e9781ed164aaf105a7880":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-07 20:00:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/47/701447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/701447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/701447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/701447/2"}}},"commit":{"parents":[{"commit":"21c5a7d6cb37e78f2c141157f4fd738b37b95b95","subject":"Retire x/dox (step 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21c5a7d6cb37e78f2c141157f4fd738b37b95b95"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-07 19:55:21.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-07 19:59:43.000000000","tz":60},"subject":"Finish retiring x/dox (step 4)","message":"Finish retiring x/dox (step 4)\n\nRemove x/dox from OpenDev and mark it as retired.\n\nDepends-On: https://review.opendev.org/701443\nDepends-On: https://review.opendev.org/701448\nChange-Id: Ic8320d4190eaacee28c7cdcd94cadf2b66313af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/185408332dae90a7f67e9781ed164aaf105a7880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/185408332dae90a7f67e9781ed164aaf105a7880"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
