)]}'
{"id":"openstack%2Fproject-config~701509","triplet_id":"openstack%2Fproject-config~master~I5e1a56ee32a1fce4eaa4ecddc167f7c68405335a","project":"openstack/project-config","branch":"master","topic":"retire-js-openstack-lib","hashtags":[],"change_id":"I5e1a56ee32a1fce4eaa4ecddc167f7c68405335a","subject":"Retire js-openstack-lib (step 2)","status":"ABANDONED","created":"2020-01-08 08:42:01.000000000","updated":"2020-01-10 09:51:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39bfa4b3f8f1fa993e59564cf889acd809bd085c","_number":701509,"virtual_id_number":701509,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-01-10 08:29:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-08 08:49:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-01-08 09:27:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 08:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-10 08:29:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"96fd1e8cf7579303e2e0ea3f8c30b8c51d1cde11","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 08:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13a0d21b74cec06a16f6a5f85bd5b6de5b410a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 08:49:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26906f2d94684bf99336415a3394677f : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"46fde12fd39df9da50397c1a5487e1f4d3957d08","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 09:27:05.000000000","message":"Patch Set 1: Workflow-1\n\nSeems there\u0027s interest to keep it...","accounts_in_message":[],"_revision_number":1},{"id":"35b062dd2b904a6a53044917da06a2d0c15d7fdf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-08 09:34:48.000000000","message":"Patch Set 1:\n\nThe interest is /me","accounts_in_message":[],"_revision_number":1},{"id":"6497351341da38bad7f8f7c84a2de8092c2cd5b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 08:28:49.000000000","message":"Patch Set 1:\n\nproject changes governance: https://review.opendev.org/701854","accounts_in_message":[],"_revision_number":1},{"id":"f0c4a8a42f1ad3ce625c9bd9184d60c54bf0de53","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 08:29:20.000000000","message":"Patch Set 1:\n\nThough note I will still want to remove the jobs for now and move them all in repo.","accounts_in_message":[],"_revision_number":1},{"id":"656c548df1a5baad5e9aff35e4222c5750fa78f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-10 09:51:50.000000000","message":"Abandoned\n\nwe won\u0027t retire the repo","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b7289f63a2ca65e8b529b82041d429068c8e3a3","revisions":{"0b7289f63a2ca65e8b529b82041d429068c8e3a3":{"kind":"REWORK","_number":1,"created":"2020-01-08 08:42:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/701509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/701509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/701509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/701509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/701509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/701509/1"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-08 08:39:53.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-08 08:39:53.000000000","tz":60},"subject":"Retire js-openstack-lib (step 2)","message":"Retire js-openstack-lib (step 2)\n\nRemove jobs from openstack/js-openstack-lib repostiry\nand add noop-jobs to retire the repository.\n\nThis is step 2 from\nhttps://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nStep 1 is done, it was the announcement which is at:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-January/011860.html\n\nChange-Id: I5e1a56ee32a1fce4eaa4ecddc167f7c68405335a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b7289f63a2ca65e8b529b82041d429068c8e3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b7289f63a2ca65e8b529b82041d429068c8e3a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
