)]}'
{"id":"openstack%2Fproject-config~701530","triplet_id":"openstack%2Fproject-config~master~Ice47cf6b84cf817db47867aa536d4f4b80c9b792","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"Ice47cf6b84cf817db47867aa536d4f4b80c9b792","subject":"New repo: \"ideas\"","status":"MERGED","created":"2020-01-08 10:47:52.000000000","updated":"2020-01-15 16:16:07.000000000","submitted":"2020-01-15 16:16:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701530-1579104967808-1b68ee4e","meta_rev_id":"28db9986a1e6d08e189d81584660040e97ff06cb","_number":701530,"virtual_id_number":701530,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 16:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-15 15:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-15 15:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-15 15:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 15:55:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-15 15:57:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-15 16:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce82b06f72b42f175d819f2061abcd294d0933b5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2054577644b3e60538a50c4f59cc1286b25b2fbf","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:48:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b2baaa24bf68d840837836c3e32c82ffdb6e198f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:49:04.000000000","message":"Patch Set 2: Workflow-1\n\nWIP -- need needed-by","accounts_in_message":[],"_revision_number":2},{"id":"2c9fcbffa4ec69aca04e943a352079fa3c7bb1b1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:55:36.000000000","message":"Removed Workflow-1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"4fa40d0d87b3bf2627716e9b1de7726ca7a9ee0e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:57:09.000000000","message":"Patch Set 2:\n\nI thought I needed a \"Needed-by\", but I don\u0027t: Before updating the system-config for the governance vhost, I need to actively have a repo which can build the docs, and there is no other repo that need to be aware of this before the docs need to be built, AFAIK.","accounts_in_message":[],"_revision_number":2},{"id":"74ce5ff3c2e835cc24c20b2869b797ef3a551e97","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 10:58:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"31add8731cf9edb3f933476596fa2bacecd1d52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:12:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a7ae226538914280aaaebb3be3a42fd0 : SUCCESS in 7m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/14a37dd9ca6a42528a3a899425476c16 : SUCCESS in 6m 40s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f67a87db833749acb26594438f563fba : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0b5d98985bb0643277cec301b038e52b21d03fb0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 15:10:51.000000000","message":"Patch Set 3:\n\nthe needed-by should link to the change to the governance repo that reflects who owns this repo.\n\nI don\u0027t like the name \"Ideas\", it\u0027s where generic. Want to make it more specific? Also, please use lowercase \"i\".","accounts_in_message":[],"_revision_number":3},{"id":"1e9ad4e602478c519feb45d8cc12ba8bc1cebfd6","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 16:56:13.000000000","message":"Patch Set 3:\n\n\u003e the needed-by should link to the change to the governance repo that\n \u003e reflects who owns this repo.\n\nAFAIK, TC owns governance repo, and doesn\u0027t need to get listed itself in projects.yaml. It\u0027s just documentation.\n\n \u003e \n \u003e I don\u0027t like the name \"Ideas\", it\u0027s where generic. Want to make it\n \u003e more specific? Also, please use lowercase \"i\".\n\nGood catch on the lowercase/uppercase. For the name, it _is_ supposed to be generic, dull, and easy to remember, though :)","accounts_in_message":[],"_revision_number":3},{"id":"5647eb892e465dfd17c23d9de4ab4e419ce9ca68","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 16:57:18.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ec01102767a47f8d86f042102281a99ddde6e338","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 16:58:05.000000000","message":"Patch Set 4:\n\nJP, the repos should be added to https://opendev.org/openstack/governance/src/branch/master/reference/technical-committee-repos.yaml\n\nPlease do this with a new change and link to it via needed-by.","accounts_in_message":[],"_revision_number":4},{"id":"d8b311556196018e9ba9d9666ba91b102d422c29","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 16:58:06.000000000","message":"Patch Set 4:\n\n\u003e \u003e the needed-by should link to the change to the governance repo\n \u003e that\n \u003e \u003e reflects who owns this repo.\n \u003e \n \u003e AFAIK, TC owns governance repo, and doesn\u0027t need to get listed\n \u003e itself in projects.yaml. It\u0027s just documentation.\n \u003e \n \u003e \u003e\n \u003e \u003e I don\u0027t like the name \"Ideas\", it\u0027s where generic. Want to make\n \u003e it\n \u003e \u003e more specific? Also, please use lowercase \"i\".\n \u003e \n \u003e Good catch on the lowercase/uppercase. For the name, it _is_\n \u003e supposed to be generic, dull, and easy to remember, though :)\n\nThough I guess I could go for openstack-ideas, would that be fine for you?","accounts_in_message":[],"_revision_number":4},{"id":"0b7e554cf493e4034e1357950bf7a3ee62414d23","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 16:59:12.000000000","message":"Patch Set 4:\n\nNaming is tricky - not sure whether openstack-ideas is better than ideas. Or should it be \"cross-product-ideas\"? Any ideas ;) where to ask for good names?","accounts_in_message":[],"_revision_number":4},{"id":"766b97ef254284e58c6a2655123a651d748878b9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-08 17:00:00.000000000","message":"Patch Set 4:\n\nbtw. we have openstack-specs - what\u0027s the difference to ideas?","accounts_in_message":[],"_revision_number":4},{"id":"438a89201f8fe63c44ecf1c7d60aa1088617ee5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 17:10:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/79213807959c430ba48ff94f752f8529 : SUCCESS in 7m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f6e09b52b1d6423ca2bdcad845ac9534 : SUCCESS in 4m 10s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/68d85108c11941ab997bb0b8a6a10c63 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2832c345069d6a49a32c4fa38cd7441343777c83","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-09 07:53:00.000000000","message":"Patch Set 4:\n\n\u003e JP, the repos should be added to https://opendev.org/openstack/governance/src/branch/master/reference/technical-committee-repos.yaml\n \u003e \n \u003e Please do this with a new change and link to it via needed-by.\n\nAhah yes I forgot that one!","accounts_in_message":[],"_revision_number":4},{"id":"c5c45ae30dc4732f555f784723f469d7ade39a66","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-09 07:57:43.000000000","message":"Patch Set 4:\n\n\u003e btw. we have openstack-specs - what\u0027s the difference to ideas?\n\nopenstack-specs links to all project specs.\n\nideas is a very simple way to propose (crazy or not) ideas for OpenStack in general. It\u0027s purpose is to be a lightweight process, not like a \"spec\". Please also check this: https://github.com/evrardjp/ideas/blob/master/doc/source/index.rst .\n\nI am inclined to keep \"ideas\" -- first come, first served! :)","accounts_in_message":[],"_revision_number":4},{"id":"84e18efa22e8a13c2abf230cf56a0e8594dc9336","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-09 08:04:54.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a02a69e4d5d8824b85f34c4a0a9b84d013324e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 08:18:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b955d5f4fbd144ada546e4e8ee12bb2c : SUCCESS in 7m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b3335c99c2fb4eecaca0055861a1719e : SUCCESS in 4m 49s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3fb51ff3edd847ed9b73131a8ad874ed : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":5},{"id":"8c3293b9297f6b713135b7247267881fb4b9b9ff","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 15:55:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f19bf09ee5cd75a246f3b5f511cae1e7bfcce644","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 15:57:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"de07a33ebda95c3a7e607b8afebdbb2a28ad5e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:57:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ef6ed1b0e2f8659afbe879a405ea4a3f52a9728e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:16:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9319e71cccf9484988a4c42cd031490c : SUCCESS in 8m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b0c3e3a0363b492db858a180dedc6146 : SUCCESS in 5m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/296f75cbbbbc4b81a5c5dcd2020e58d7 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":5},{"id":"aefaab77dc7c220b6eee9c5636d2013063fcd9b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:16:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e94a0f641f76cc36133a4cb9c1a0316dee89d482","revisions":{"fb0dd0854c690659f5e29ba94d8be5b427630a77":{"kind":"REWORK","_number":1,"created":"2020-01-08 10:47:52.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/30/701530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/701530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/701530/1"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"subject":"New project \"Ideas\"","message":"New project \"Ideas\"\n\nThis is the starting point for the \"ideas\" project.\n\nIdeas is a repository where everyone can submit their ideas for\nglobal OpenStack changes. It is the follow up of a discussion\nin the last summit in Shanghai, where we discussed the lack of an easy to\nbrowse and search history of all the crazy ideas for OpenStack created in the\npast, as the ML is not really appropriate for this searching and revision\ncontrol.\n\nChange-Id: Ice47cf6b84cf817db47867aa536d4f4b80c9b792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb0dd0854c690659f5e29ba94d8be5b427630a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb0dd0854c690659f5e29ba94d8be5b427630a77"}]},"branch":"refs/heads/master"},"eb5dc09411590b8aedfca3b964043379938db35b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-08 10:48:37.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/30/701530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/701530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/701530/2"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:48:03.000000000","tz":60},"subject":"New project \"Ideas\"","message":"New project \"Ideas\"\n\nThis is the starting point for the \"ideas\" project.\n\nIdeas is a repository where everyone can submit their ideas for\nglobal OpenStack changes. It is the follow up of a discussion\nin the last summit in Shanghai, where we discussed the lack of an easy\nto browse and search history of all the crazy ideas for OpenStack\ncreated in the past, as the ML is not really appropriate for this\nsearching and revision control.\n\nChange-Id: Ice47cf6b84cf817db47867aa536d4f4b80c9b792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb5dc09411590b8aedfca3b964043379938db35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb5dc09411590b8aedfca3b964043379938db35b"}]},"branch":"refs/heads/master"},"a0121df6f8742b28db2f84de226f7e36bfa5f2d0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-08 10:58:54.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/30/701530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/701530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/701530/3"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:58:51.000000000","tz":0},"subject":"New project \"Ideas\"","message":"New project \"Ideas\"\n\nThis is the starting point for the \"ideas\" project.\nIdeas is a repository where everyone can submit their ideas for\nglobal OpenStack changes.\n\nIt is the follow up of a discussion in the last summit in Shanghai,\nwhere we discussed the lack of an easy way to browse and search history\nof all the (crazy) ideas for OpenStack floated on the mailing lists.\nWe established (due to recurring topics) that the MLs are not really\nappropriate for this searching and revision control.\n\nChange-Id: Ice47cf6b84cf817db47867aa536d4f4b80c9b792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a0121df6f8742b28db2f84de226f7e36bfa5f2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a0121df6f8742b28db2f84de226f7e36bfa5f2d0"}]},"branch":"refs/heads/master"},"8136a84144c6d1ced74386d32300149874e74117":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-08 16:57:18.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/30/701530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/701530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/701530/4"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 16:57:15.000000000","tz":0},"subject":"New repo: \"ideas\"","message":"New repo: \"ideas\"\n\nThis is the starting point for the \"ideas\" project.\nIdeas is a repository where everyone can submit their ideas for\nglobal OpenStack changes.\n\nIt is the follow up of a discussion in the last summit in Shanghai,\nwhere we discussed the lack of an easy way to browse and search history\nof all the (crazy) ideas for OpenStack floated on the mailing lists.\nWe established (due to recurring topics) that the MLs are not really\nappropriate for this searching and revision control.\n\nChange-Id: Ice47cf6b84cf817db47867aa536d4f4b80c9b792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8136a84144c6d1ced74386d32300149874e74117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8136a84144c6d1ced74386d32300149874e74117"}]},"branch":"refs/heads/master"},"e94a0f641f76cc36133a4cb9c1a0316dee89d482":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-01-09 08:04:54.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/30/701530/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/701530/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/701530/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/701530/5"}}},"commit":{"parents":[{"commit":"b025023dd2439dedfca1f6fc095cdf737f481f8f","subject":"Disable compression of files in stage-output role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b025023dd2439dedfca1f6fc095cdf737f481f8f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-08 10:41:34.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-09 08:04:51.000000000","tz":0},"subject":"New repo: \"ideas\"","message":"New repo: \"ideas\"\n\nThis is the starting point for the \"ideas\" project.\nIdeas is a repository where everyone can submit their ideas for\nglobal OpenStack changes.\n\nIt is the follow up of a discussion in the last summit in Shanghai,\nwhere we discussed the lack of an easy way to browse and search history\nof all the (crazy) ideas for OpenStack floated on the mailing lists.\nWe established (due to recurring topics) that the MLs are not really\nappropriate for this searching and revision control.\n\nNeeded-By: https://review.opendev.org/#/c/701678\nChange-Id: Ice47cf6b84cf817db47867aa536d4f4b80c9b792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e94a0f641f76cc36133a4cb9c1a0316dee89d482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e94a0f641f76cc36133a4cb9c1a0316dee89d482"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
