)]}'
{"id":"openstack%2Fproject-config~701982","triplet_id":"openstack%2Fproject-config~master~Ibe1f0fba6ae2a459be22b33d8de4285b739a1df0","project":"openstack/project-config","branch":"master","topic":"check-release-approval","hashtags":[],"change_id":"Ibe1f0fba6ae2a459be22b33d8de4285b739a1df0","subject":"Define check-release-approval executor job","status":"MERGED","created":"2020-01-10 15:33:40.000000000","updated":"2020-01-23 17:17:24.000000000","submitted":"2020-01-23 17:17:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701982-1579799844396-c94174cc","meta_rev_id":"6290b2d9bbd1ae9597f54c4018541c8f3e7db975","_number":701982,"virtual_id_number":701982,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 17:17:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-01-23 14:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-23 17:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-23 17:04:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 14:59:37.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-01-23 17:04:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-23 17:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0d66ba4fc45526cd48af7adf1375d81aae2b56d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-10 15:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"628d9438b13aaf216ad41a0704dd1b5203ced460","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-10 15:38:20.000000000","message":"Patch Set 1: Workflow-1\n\nHmm, let me put governance in required-projects instead of fetching projects.yaml over the network...","accounts_in_message":[],"_revision_number":1},{"id":"b4dc11784d44dcd4c6ac586856e8b6dd85eea6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 15:49:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42a7941b0e2543edb25b0e045d77bfd8 : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6d087ec8105a00e5392cfc58c8f4ffdeabe28240","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-10 15:55:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53f3a5acdb659032ee7c3d7dd04e7225b28e4b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a77c0fbd2904669bcad6a689c55f451 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9ab1b8a0764ecf59b81817d07ef63bf007eb4ddc","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-01-23 14:59:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"472c488d97e0055ce4eba53edad96eeb46e5b4b3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-23 17:04:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nSuggestion inline you might consider for a future iteration.","accounts_in_message":[],"_revision_number":2},{"id":"ffce915afee0e423daacb63f38d958cfcd49b293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:04:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f12161e553783868c78f214093474ae5b858884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:17:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd5a32b5d290ee9beed80891bfd8eba311450c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:17:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0bc0b4e80cf34b1fb60c1192ea04dee1 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7be4b337bb10ea33e15c9e4a39fc5b7af12c06e0","revisions":{"a22fe978ad49b4a360592ce51c4ba97bf2156f7b":{"kind":"REWORK","_number":1,"created":"2020-01-10 15:33:40.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/82/701982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/701982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/701982/1"}}},"commit":{"parents":[{"commit":"bfe28f2601812c3e6330b9317bb5e9f3d556885b","subject":"Merge \"Add periodic panel to Octavia dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfe28f2601812c3e6330b9317bb5e9f3d556885b"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-10 14:42:56.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-10 15:30:29.000000000","tz":60},"subject":"Define check-release-approval executor job","message":"Define check-release-approval executor job\n\nAs discussed in [1], create a job that will check that a release request\nhas been approved by the PTL or a release liaison attached to the\ndeliverable.\n\nSince this job is very short and will be rerun every time a vote is\nposted, rather than consume a remote host, it will be run directly on\nthe executor. To that effect, the python script called only uses stdlib\nor modules that are already present on the executor (requests and yaml).\n\n[1] http://lists.openstack.org/pipermail/openstack-infra/2019-December/006556.html\n\nChange-Id: Ibe1f0fba6ae2a459be22b33d8de4285b739a1df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a22fe978ad49b4a360592ce51c4ba97bf2156f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a22fe978ad49b4a360592ce51c4ba97bf2156f7b"}]},"branch":"refs/heads/master"},"7be4b337bb10ea33e15c9e4a39fc5b7af12c06e0":{"kind":"REWORK","_number":2,"created":"2020-01-10 15:55:27.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/82/701982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/701982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/701982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/701982/2"}}},"commit":{"parents":[{"commit":"bfe28f2601812c3e6330b9317bb5e9f3d556885b","subject":"Merge \"Add periodic panel to Octavia dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfe28f2601812c3e6330b9317bb5e9f3d556885b"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-10 14:42:56.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-10 15:52:35.000000000","tz":60},"subject":"Define check-release-approval executor job","message":"Define check-release-approval executor job\n\nAs discussed in [1], create a job that will check that a release request\nhas been approved by the PTL or a release liaison attached to the\ndeliverable.\n\nSince this job is very short and will be rerun every time a vote is\nposted, rather than consume a remote host, it will be run directly on\nthe executor. To that effect, the python script called only uses stdlib\nor modules that are already present on the executor (requests and yaml).\n\n[1] http://lists.openstack.org/pipermail/openstack-infra/2019-December/006556.html\n\nChange-Id: Ibe1f0fba6ae2a459be22b33d8de4285b739a1df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7be4b337bb10ea33e15c9e4a39fc5b7af12c06e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7be4b337bb10ea33e15c9e4a39fc5b7af12c06e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
