)]}'
{"id":"openstack%2Fproject-config~702028","triplet_id":"openstack%2Fproject-config~master~Ied4d243fd840f57301ab620c9a75a243bdc61eae","project":"openstack/project-config","branch":"master","topic":"gerritbot","hashtags":[],"change_id":"Ied4d243fd840f57301ab620c9a75a243bdc61eae","subject":"Remove retired x/js-* repos from gerritbot","status":"MERGED","created":"2020-01-10 19:05:28.000000000","updated":"2020-01-16 17:38:59.000000000","submitted":"2020-01-16 17:38:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702028-1579196339978-f3637170","meta_rev_id":"1e535fb9ed443d641c76346854239122baae6468","_number":702028,"virtual_id_number":702028,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 17:38:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2020-01-16 15:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2020-01-16 17:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-16 16:55:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2020-01-16 17:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 15:44:42.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2020-01-16 16:55:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-16 17:26:59.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-16 17:38:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c704299832945287460c6613f000af1465088912","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-10 19:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"416d762eca6004543578bd60efcde421fb1a7222","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 19:09:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9a2bcf6fab46e4e756994bd84853f670da1f65d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 19:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/456c14754c2f4f0d84bca57231308c76 : SUCCESS in 6m 32s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c3ed98688cd0424e86014a8c3d419bd6 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6987009b2a24e686cde041de577621980904f0fa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 15:24:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9016ec5ad768c72b0544e0c5b2893eb4c7306d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 15:37:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/10e9437e191f4b87af034edf4e99ccbd : SUCCESS in 8m 15s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c3c82b3d4ffb4c99a036abdb89b99a02 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d68c46794b3bd59172a0a1c808af367e4650628c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-01-16 15:44:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bdb556b58fb9656e5c17ccf2332a51d8ee8942ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-16 16:55:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"38a45102ed84c296ef4163f8fb54ebdfcf4ebc56","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-16 17:26:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0aa5779d61b634d052979e43f17a72e8543ea7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:27:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e783a3c1902982a75b90204573ba4a9147e3f9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:38:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"83b2eefbc32c2e0bf5bc0a451093f01288cc3ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:38:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9062672bbf714a4dbccc140d1025ee13 : SUCCESS in 8m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/272ed54e3e5244c9916ef33a82919516 : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e695eeb6470fc8d4bc298e7d6137aa49b9029d3","revisions":{"0e548e171bf9da9bacb6c22df9599da65e8b1f50":{"kind":"REWORK","_number":1,"created":"2020-01-10 19:05:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/28/702028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/702028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/702028/1"}}},"commit":{"parents":[{"commit":"da628d77bc64c535f2785f894204d6043475450b","subject":"Merge \"Remove duplicated ACL files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da628d77bc64c535f2785f894204d6043475450b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-10 19:03:51.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-10 19:05:15.000000000","tz":60},"subject":"Remove retired x/js-* repos from gerritbot","message":"Remove retired x/js-* repos from gerritbot\n\nRemove the following gerritbot notifications since those repos have\nbeen retired already but gerritbot was not updated:\n- x/js-afs-blob-store\n- x/js-generator-openstack\n- x/js-openstack-registry-hooks\n\nChange-Id: Ied4d243fd840f57301ab620c9a75a243bdc61eae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e548e171bf9da9bacb6c22df9599da65e8b1f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0e548e171bf9da9bacb6c22df9599da65e8b1f50"}]},"branch":"refs/heads/master"},"7e695eeb6470fc8d4bc298e7d6137aa49b9029d3":{"kind":"REWORK","_number":2,"created":"2020-01-16 15:24:47.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/28/702028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/702028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/702028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/702028/2"}}},"commit":{"parents":[{"commit":"c0a35735b67c882f3ba6c1612e4a1d3a28e397bb","subject":"Merge \"Remove legacy-gearman-plugin-mavin-build-ubuntu-trusty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0a35735b67c882f3ba6c1612e4a1d3a28e397bb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-10 19:03:51.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-16 15:24:35.000000000","tz":60},"subject":"Remove retired x/js-* repos from gerritbot","message":"Remove retired x/js-* repos from gerritbot\n\nRemove the following gerritbot notifications since those repos have\nbeen retired already but gerritbot was not updated:\n- x/js-afs-blob-store\n- x/js-generator-openstack\n- x/js-openstack-registry-hooks\n\nChange-Id: Ied4d243fd840f57301ab620c9a75a243bdc61eae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e695eeb6470fc8d4bc298e7d6137aa49b9029d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e695eeb6470fc8d4bc298e7d6137aa49b9029d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
