)]}'
{"id":"openstack%2Fproject-config~702071","triplet_id":"openstack%2Fproject-config~master~Ic526595e4eac0e530eeb7603a88165b9616e0d17","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ic526595e4eac0e530eeb7603a88165b9616e0d17","subject":"New charm for watcher","status":"MERGED","created":"2020-01-11 10:24:59.000000000","updated":"2020-01-15 21:07:26.000000000","submitted":"2020-01-15 15:01:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702071-1579100514274-bbef5f2b","meta_rev_id":"9f28ebf755bbf7b3b4c8832e4b501700b8ba3942","_number":702071,"virtual_id_number":702071,"owner":{"_account_id":28014,"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","username":"skatsaounis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 15:01:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-01-13 16:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-01-13 08:47:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2020-01-13 10:40:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"},{"value":2,"date":"2020-01-15 14:38:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"date":"2020-01-15 21:07:26.000000000","post_submit":true,"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"},{"value":1,"date":"2020-01-15 14:38:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 08:47:42.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2020-01-13 10:40:06.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2020-01-13 16:08:41.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-15 14:38:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-01-15 15:01:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-15 21:07:26.000000000","updated_by":{"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"},"reviewer":{"_account_id":30442,"name":"Alexandros Soumplis","email":"soumplis@admin.grnet.gr","username":"soumplis"},"state":"REVIEWER"}],"messages":[{"id":"6c525008fb40c9df488a5cde51688977dc6074c3","author":{"_account_id":28014,"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","username":"skatsaounis"},"date":"2020-01-11 10:24:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ceb59df8d61893a3cded579449edf136b541cff6","author":{"_account_id":28014,"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","username":"skatsaounis"},"date":"2020-01-11 10:28:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3bd4d61c8c90e0c3213681af8eee29123829530a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 10:40:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/45e446587244489497d2b3c8daf564da : SUCCESS in 6m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e35c626908da4b3a9e2e0aa68939e3e8 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"032d51246491d0c03f718c2c79cd8f9c7296b2c7","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-01-13 08:47:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bce6a75434631d51b5e873e9f47037a66a52f304","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-01-13 10:40:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4b571361cc00d33c473b22ae49e4b21046461e4a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-13 16:08:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9066b6f9f79c8db7c101ee5c4099b2b2b9a16434","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-15 14:38:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d51dbebf0202dbf6713222aef655a057e386bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 14:38:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1bdb46476a93f223f64acbae4e72bfd5ed10a87d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:01:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1ad8f38ddf96e0eb5a65633f1636e8095a2bff25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:01:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e90d34c276a747f99c4e135a1f074a6f : SUCCESS in 8m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c426ac3f0cbc40078ab195b9b4c0463b : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a54ab1311697287960ab6cf4ad4b761e83fd216","revisions":{"3ed8127854b97cccc232097534237eb75994ece7":{"kind":"REWORK","_number":1,"created":"2020-01-11 10:24:59.000000000","uploader":{"_account_id":28014,"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","username":"skatsaounis"},"ref":"refs/changes/71/702071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/702071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/702071/1"}}},"commit":{"parents":[{"commit":"da628d77bc64c535f2785f894204d6043475450b","subject":"Merge \"Remove duplicated ACL files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da628d77bc64c535f2785f894204d6043475450b"}]}],"author":{"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","date":"2020-01-11 10:09:37.000000000","tz":120},"committer":{"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","date":"2020-01-11 10:24:56.000000000","tz":120},"subject":"New charm for watcher","message":"New charm for watcher\n\nAdd watcher charm to support charmed OpenStack Watcher\n\nChange-Id: Ic526595e4eac0e530eeb7603a88165b9616e0d17\nSigned-off-by: Stamatis Katsaounis \u003cskatsaounis@admin.grnet.gr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ed8127854b97cccc232097534237eb75994ece7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ed8127854b97cccc232097534237eb75994ece7"}]},"branch":"refs/heads/master"},"4a54ab1311697287960ab6cf4ad4b761e83fd216":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-11 10:28:50.000000000","uploader":{"_account_id":28014,"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","username":"skatsaounis"},"ref":"refs/changes/71/702071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/702071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/702071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/702071/2"}}},"commit":{"parents":[{"commit":"da628d77bc64c535f2785f894204d6043475450b","subject":"Merge \"Remove duplicated ACL files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da628d77bc64c535f2785f894204d6043475450b"}]}],"author":{"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","date":"2020-01-11 10:09:37.000000000","tz":120},"committer":{"name":"Stamatis Katsaounis","email":"skatsaounis@admin.grnet.gr","date":"2020-01-11 10:27:59.000000000","tz":120},"subject":"New charm for watcher","message":"New charm for watcher\n\nAdd watcher charm to support charmed OpenStack Watcher\n\nNeeded-By: Iba68b49c1eedd29afaca67bc3c2ed2f716da7a57\nChange-Id: Ic526595e4eac0e530eeb7603a88165b9616e0d17\nSigned-off-by: Stamatis Katsaounis \u003cskatsaounis@admin.grnet.gr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a54ab1311697287960ab6cf4ad4b761e83fd216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4a54ab1311697287960ab6cf4ad4b761e83fd216"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
