)]}'
{"id":"openstack%2Fproject-config~702259","triplet_id":"openstack%2Fproject-config~master~Ia23ee2e1a769d17c6c486bcf4d192897e98e086a","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ia23ee2e1a769d17c6c486bcf4d192897e98e086a","subject":"x/cloudbase-init: remove openstack gates","status":"MERGED","created":"2020-01-13 17:02:01.000000000","updated":"2020-01-15 15:37:05.000000000","submitted":"2020-01-15 15:37:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702259-1579102625460-fa858c1e","meta_rev_id":"7e063e5f234b3091811b8efe564840a9778af736","_number":702259,"virtual_id_number":702259,"owner":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-01-14 10:11:22.000000000","_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},{"value":0,"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 15:37:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"all":[{"value":1,"date":"2020-01-13 18:11:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},{"value":1,"date":"2020-01-14 13:03:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"value":2,"date":"2020-01-13 17:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-15 15:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},{"value":0,"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-15 15:24:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 17:46:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-13 18:11:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-14 13:03:26.000000000","updated_by":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"reviewer":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"state":"REVIEWER"},{"updated":"2020-01-15 15:24:01.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-15 15:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df2492c82ca2d4a10084ca307b11ab3b7a6ad36f","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2020-01-13 17:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"681a581204f5ceea70108dac6746a5e6db3cc6a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 17:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a2fdafb7f124e849570e9c17fb998ae : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ebfdf65918247676de13023d8d1e63c8f0d6013c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-13 17:11:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b33d1f65eb96d7b03ace0ed4c1158090e52a7ca5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-13 17:13:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50944023eee77d92365923c72dbc02c4bda13587","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2020-01-13 17:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dd439e591183e3ea8f33d67ef048a1e7be0b62c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-13 17:46:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a18ce606f76ba9c883222331232f083f70c453c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 17:46:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18bd7351b23449a4bc33aa506bff5e15 : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":2},{"id":"400b5f66ce3b1200987bd0dd8009d303f14e1751","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-13 18:11:30.000000000","message":"Patch Set 2: Code-Review+1\n\nSomething for the project, but if you are no longer testing py27, then declaring py27 support in setup.cfg should be removed. Not testing \u003d\u003d broken.","accounts_in_message":[],"_revision_number":2},{"id":"f5ad5ae084c3cd5c11095b48b1177eb5add7f795","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2020-01-14 10:11:22.000000000","message":"Patch Set 2:\n\nSean, I will keep the py27 gates for the moment, in cloudbase-init zuul config.","accounts_in_message":[],"_revision_number":2},{"id":"06fe45e6737e605695fb6560104858018b7490bf","author":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"date":"2020-01-14 13:03:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3acbecb1c8521dda5584b3b2c35ff42197ca84f5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-15 15:24:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"66d23124a528bf8379f25e4efa7fc9696a584fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:24:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b787d568f7b15d79adb78c3e06a080996a98c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:37:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ca8313f219543228c5aca92fa0c49ea : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"414c4ee8384b20e59a4ed9d0615f4da137eb3477","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:37:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52a522acb9587f9bdf97a98e06daaaaff62a2ca1","revisions":{"9eb5ec830c1577bdbeb201b3d615de9c4951f0c4":{"kind":"REWORK","_number":1,"created":"2020-01-13 17:02:01.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/59/702259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/702259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/702259/1"}}},"commit":{"parents":[{"commit":"abc91b7cedb9f22370748a5b41afe0f807962ca3","subject":"Don\u0027t mirror wheels for pip/setuptools/virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abc91b7cedb9f22370748a5b41afe0f807962ca3"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2020-01-13 17:00:27.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2020-01-13 17:01:59.000000000","tz":120},"subject":"cloudbase-init: set zuul project to noop-jobs","message":"cloudbase-init: set zuul project to noop-jobs\n\nLet cloudbase-init\u0027s .zuul.yml to define the zuul gates.\n\nChange-Id: Ia23ee2e1a769d17c6c486bcf4d192897e98e086a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9eb5ec830c1577bdbeb201b3d615de9c4951f0c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9eb5ec830c1577bdbeb201b3d615de9c4951f0c4"}]},"branch":"refs/heads/master"},"52a522acb9587f9bdf97a98e06daaaaff62a2ca1":{"kind":"REWORK","_number":2,"created":"2020-01-13 17:35:26.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/59/702259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/702259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/702259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/702259/2"}}},"commit":{"parents":[{"commit":"abc91b7cedb9f22370748a5b41afe0f807962ca3","subject":"Don\u0027t mirror wheels for pip/setuptools/virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abc91b7cedb9f22370748a5b41afe0f807962ca3"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2020-01-13 17:00:27.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2020-01-13 17:33:16.000000000","tz":120},"subject":"x/cloudbase-init: remove openstack gates","message":"x/cloudbase-init: remove openstack gates\n\nLet cloudbase-init\u0027s .zuul.yml define the zuul gates.\nThe publish-to-pypi template can be left as-is, so that\nthe pypi package gets updated on new tags/releases.\n\nChange-Id: Ia23ee2e1a769d17c6c486bcf4d192897e98e086a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52a522acb9587f9bdf97a98e06daaaaff62a2ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52a522acb9587f9bdf97a98e06daaaaff62a2ca1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
