)]}'
{"id":"openstack%2Fproject-config~702310","triplet_id":"openstack%2Fproject-config~master~I0a7d6f93f07477b6dd29ab143130dd9064c250be","project":"openstack/project-config","branch":"master","topic":"pep503-indexes","hashtags":[],"change_id":"I0a7d6f93f07477b6dd29ab143130dd9064c250be","subject":"Update hacking, fix errors/warnings","status":"MERGED","created":"2020-01-13 23:49:30.000000000","updated":"2020-01-14 00:34:07.000000000","submitted":"2020-01-14 00:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702310-1578961959789-1b68ee4e","meta_rev_id":"fec9e3f774796a24e8a4b0ae230c5e18a651d91a","_number":702310,"virtual_id_number":702310,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 00:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-13 23:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-13 23:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-13 23:57:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 23:55:56.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-13 23:57:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-14 00:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2325b9f990001d561d9a0137d1ea25dc52514ce6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-13 23:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9372b6466df41633fe3d1e5d478a5704a3215aab","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-13 23:55:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"babe36db33c8ae3b08f88b9da97914401d0fff93","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-13 23:57:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b740c477f64199f87e771302b134f7bc161ff497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:24:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/9ab89210906c41f28b5179a670e025fc : SUCCESS in 2m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/95015b05befe406ab3ae5a245af5e9e6 : SUCCESS in 5m 43s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6b3c8d04ffe54ffe89933debe3a9b70c : SUCCESS in 3m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2552ce2be6a946a983370bb6652bbfd6 : SUCCESS in 3m 49s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/7cbd9c6a535d48d4936d8fee404bf19e : SUCCESS in 4m 05s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/7b47c0d30a374b15bf339c8154443817 : SUCCESS in 3m 07s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e8d35a91558447ac822900b150334869 : SUCCESS in 7m 00s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/01701292e497475fa4d5d2545f0ddcd2 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"20878f18a66cfaea607021be03ed7e911267fc70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:24:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db777dae8dd90335ad3b0c34804d0adbd9200206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:32:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/40949694c9c14ac188f7fed2c4442796 : SUCCESS in 3m 21s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e80cc8cdab2453c93ba473e03675354 : SUCCESS in 6m 28s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/2bb110edfbe748cc87d9a34f179fbb93 : SUCCESS in 3m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/606ca90776bc495b9cccf1dc7ccb2a6c : SUCCESS in 3m 49s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0676eec8b2e4457798e17249c0b432a7 : SUCCESS in 3m 20s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/283fc82a38b64511ad43875856decf43 : SUCCESS in 3m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d831ca9589064d25bb3af200c63644c0 : SUCCESS in 6m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e3eadb1b287a4906878834ff40e1aa19 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"28e31ca2c4c9f7f26f435092ba2f0d9055d09508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:32:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9cca21cafd2f47112b95400bc7f64616a022b1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:34:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/93dd68c0f9ec4d40877f819b2379a93a : SUCCESS in 1m 09s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/77f1dd901ece4519844e8a0699574a00 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f6629021e085e1c2c477deb1047e90a7ecd4c86","revisions":{"4f6629021e085e1c2c477deb1047e90a7ecd4c86":{"kind":"REWORK","_number":1,"created":"2020-01-13 23:49:30.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/10/702310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/702310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/702310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/702310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/702310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/702310/1"}}},"commit":{"parents":[{"commit":"abc91b7cedb9f22370748a5b41afe0f807962ca3","subject":"Don\u0027t mirror wheels for pip/setuptools/virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abc91b7cedb9f22370748a5b41afe0f807962ca3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-13 22:39:30.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-13 22:40:45.000000000","tz":660},"subject":"Update hacking, fix errors/warnings","message":"Update hacking, fix errors/warnings\n\nThis version of hacking doesn\u0027t understand f-strings as usable in\nPython 3.  Update to the latest and fix current issues, which are all\njust formatting fixes.\n\nChange-Id: I0a7d6f93f07477b6dd29ab143130dd9064c250be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f6629021e085e1c2c477deb1047e90a7ecd4c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f6629021e085e1c2c477deb1047e90a7ecd4c86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
