)]}'
{"id":"openstack%2Fproject-config~702350","triplet_id":"openstack%2Fproject-config~master~I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc","project":"openstack/project-config","branch":"master","topic":"pep503-indexes","hashtags":[],"change_id":"I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc","subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","status":"MERGED","created":"2020-01-14 05:45:03.000000000","updated":"2020-01-16 21:53:03.000000000","submitted":"2020-01-16 21:53:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702350-1579211583517-af565c7f","meta_rev_id":"3577d4e0805758dc96a33c3c97f409fbadb1be0c","_number":702350,"virtual_id_number":702350,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 21:53:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-14 22:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-14 22:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-01-16 21:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 07:04:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-16 21:45:09.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-16 21:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"297f5dca9093d19cbbe75e48e6ff567e2a6be836","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 05:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e777062298f2ebb7397071b9c93384818bad9c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 06:04:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/006932c18da642f7918aa8d08cc5834a : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"334bd3e3bb6205c4c9e8ee272d172ea9f33ff714","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 06:24:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"75847e5eb546f7455bde4b12b6629e4c8c8a280d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 06:44:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d6ee524b0a054580a0d1155fc84354bf : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"55b5bf6afda27e434edede4c5c1ee522e1238219","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-14 07:04:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"618ed348b5eebb28e9bb14a4793855df5f056140","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 21:01:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"89648830f145c25e93fce97a634b188b63c6d4e3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 21:04:57.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"db6a2cce836a8fe2618a2423a8e81830375385ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 21:17:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9c01798a9b004c09915c4711a5cef329 : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9fd4fcefdbdc97c015e1b84b8d486b29532ff3bd","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-14 21:37:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b8e5f1b87f51b208cd9c27477e23e2f145fb069b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 21:46:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a5dc82d6affa19d18a88fe1fd9f137253d10ccee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 21:59:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fcefeb96231840a1b8b60f11a49b0fc6 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":5},{"id":"ee8efd801d7e2eb89a2e0090389dead72304386d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-14 22:28:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"47acda9dae091f0ec1ef4b604d7dd69a2fa52a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 22:41:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/67ae61c2b98846fc990f31460d1f3b86 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":6},{"id":"3caefe65884edfceeeff0916e12ffb2f65054ae7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-16 21:45:09.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"33fc19ff83a0bec8794e4e803e67c22b69ffc874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:45:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"24506ac5afcda88ccf85c471e7725b612474d902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:53:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9c5de223be1fee629b7513c5af82232de1e62f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:53:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ec8b6773efe445a8e54ceb996981f2d : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3f3fa89d0ad4e7fa812a00407860a9559d097479","revisions":{"eb9677bbe6acc14c9c347831034fa1d03803e1f7":{"kind":"REWORK","_number":1,"created":"2020-01-14 05:45:03.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/1"}}},"commit":{"parents":[{"commit":"93437f870943fc8fceb233598d42e32fac364de6","subject":"Add centos8 openafs client package build to periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93437f870943fc8fceb233598d42e32fac364de6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:44:45.000000000","tz":660},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb9677bbe6acc14c9c347831034fa1d03803e1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb9677bbe6acc14c9c347831034fa1d03803e1f7"}]},"branch":"refs/heads/master"},"0ebadc8059caf0193f0d6c55534dabe6da3b3b00":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-14 06:24:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/2"}}},"commit":{"parents":[{"commit":"1b3d77d99b2002c0c6bbaafd8d6adadb8ac35bbc","subject":"Add centos8 openafs client package build to periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1b3d77d99b2002c0c6bbaafd8d6adadb8ac35bbc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 06:24:05.000000000","tz":660},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ebadc8059caf0193f0d6c55534dabe6da3b3b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ebadc8059caf0193f0d6c55534dabe6da3b3b00"}]},"branch":"refs/heads/master"},"0467cd0eb88756728091e41d94c628cf63026c06":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-14 21:01:57.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/3"}}},"commit":{"parents":[{"commit":"2a529b6e51bd6e8e148cd38a6aebf0d21745745a","subject":"Add centos8 openafs client package build to periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a529b6e51bd6e8e148cd38a6aebf0d21745745a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 21:01:46.000000000","tz":660},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0467cd0eb88756728091e41d94c628cf63026c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0467cd0eb88756728091e41d94c628cf63026c06"}]},"branch":"refs/heads/master"},"025f39586176d925047e2aef1d1556e50e5a8af1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-14 21:04:57.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/4"}}},"commit":{"parents":[{"commit":"b276a755b7457140b68d35a795d64b7da070c039","subject":"Add centos8 openafs client package build to periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b276a755b7457140b68d35a795d64b7da070c039"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 21:04:42.000000000","tz":660},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/025f39586176d925047e2aef1d1556e50e5a8af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/025f39586176d925047e2aef1d1556e50e5a8af1"}]},"branch":"refs/heads/master"},"aa312dec590a329f1a3f5fd98497a34d1c4c3011":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-14 21:46:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/5"}}},"commit":{"parents":[{"commit":"031673aa9858f185fd4f3189bada813a356e2fd6","subject":"Add centos8 openafs client package build to periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/031673aa9858f185fd4f3189bada813a356e2fd6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 21:24:20.000000000","tz":660},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa312dec590a329f1a3f5fd98497a34d1c4c3011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa312dec590a329f1a3f5fd98497a34d1c4c3011"}]},"branch":"refs/heads/master"},"3f3fa89d0ad4e7fa812a00407860a9559d097479":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-01-14 22:28:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/702350/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/702350/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/702350/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/702350/6"}}},"commit":{"parents":[{"commit":"dc3289235e9597697f1dc21cf7a6e19ae82c7199","subject":"copy-wheels: ensure split actually splits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc3289235e9597697f1dc21cf7a6e19ae82c7199"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 05:23:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-14 22:28:06.000000000","tz":0},"subject":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs","message":"Add \u003cbuild|publish\u003e-wheel-mirror-centos-8 jobs\n\nChange-Id: I4bff59d8d2b1ef8ea96ac1211774e1a01a536fbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f3fa89d0ad4e7fa812a00407860a9559d097479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f3fa89d0ad4e7fa812a00407860a9559d097479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
