)]}'
{"id":"openstack%2Fproject-config~702572","triplet_id":"openstack%2Fproject-config~master~Id34e308509980685ed156fc68fb65975dbd0b780","project":"openstack/project-config","branch":"master","topic":"hotfix/remove_openstackid_legacy_jobs","hashtags":[],"change_id":"Id34e308509980685ed156fc68fb65975dbd0b780","subject":"Removed OpenstackId legacy jobs","status":"MERGED","created":"2020-01-15 02:23:58.000000000","updated":"2020-01-16 04:44:34.000000000","submitted":"2020-01-16 04:44:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702572-1579149874999-bbef5f2b","meta_rev_id":"02a97392a6c1244e7e872502fd540423dd96938b","_number":702572,"virtual_id_number":702572,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-01-16 04:26:53.000000000","_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 04:44:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-16 00:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-15 16:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 04:00:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 02:27:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-15 16:54:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-16 00:14:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-16 04:00:37.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-01-16 04:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcdd987b7670115af598fbe0ee98aa4606093de6","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-15 02:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5914001eaa8dcaa8a16273c84b56dc9e9d3bf44e","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-15 02:30:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e7ce7bbe43f9fa1512a5401261670ec3999ce63","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-15 02:35:20.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8d90681346fef6215f03123c802b62fc974b3662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 02:49:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c37c13c012114c6992dec4a2153f0466 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b5c5f1682f6400df7432e2572afe972db0d4458c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 07:17:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8d0f2922841d3e9f77645c30cadbd48130595ba","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-15 16:38:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"afdbe7885b8f3347ac5a7ca115d46c8ddd9e30c9","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-15 16:38:37.000000000","message":"Patch Set 3:\n\n(1 comment)\n\ndone","accounts_in_message":[],"_revision_number":3},{"id":"3f62742ab76f8ff75030f4dd49445da3fd14c660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:48:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b48b7534b1334fd38afca275629b7772 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":4},{"id":"79aa015e940a0fc040880ec7073cf8c73c21358e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 16:54:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8e3a6fe49bea3c8a4b98fd3f172f354c7e12a095","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-16 00:14:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5c1d4e17ddeb992fc865070899518dd294147d3a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-16 04:00:37.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7a3ebfd26d9165fcc50bebb435d352a68b389c3","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-16 04:26:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"70b53f22acd1e45a55b1c8912778e29973c7c965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:37:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa85ee4f360c44e5b6b66e5e0b4231fa : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"4938928ee6306b4121a803b3b05bc6451ada55a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:38:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e1a5cd77c59e8c36d827a0bf9d2718a01c9c3fb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:44:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ecde4848d25fff18ef06b6f3417d7967451c1086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:44:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9ec02769ae147cfb857cdd62c4ad779 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d8c3354749e6f36951b3b38b3c65eebb0b3ca4a","revisions":{"8923e244036dde239887eb96f1afba1051b08736":{"kind":"REWORK","_number":1,"created":"2020-01-15 02:23:58.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/72/702572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/702572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/702572/1"}}},"commit":{"parents":[{"commit":"f7aad7b27dc425d00416aa75b6b41891f75c35dd","subject":"Merge \"copy-wheels : generate indexes version 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7aad7b27dc425d00416aa75b6b41891f75c35dd"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:22:04.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:23:55.000000000","tz":-180},"subject":"Removed OpenstackId legacy jobs","message":"Removed OpenstackId legacy jobs\n\nremoved legacy jobs for openstackid project and pointed\nto new migrated zuul v3 jobs\n\nDepends-On: https://review.opendev.org/702571\nChange-Id: Id34e308509980685ed156fc68fb65975dbd0b780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8923e244036dde239887eb96f1afba1051b08736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8923e244036dde239887eb96f1afba1051b08736"}]},"branch":"refs/heads/master"},"ff3c2157431e43bcf75815144debafbbcce9761b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-15 02:30:21.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/72/702572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/702572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/702572/2"}}},"commit":{"parents":[{"commit":"f7aad7b27dc425d00416aa75b6b41891f75c35dd","subject":"Merge \"copy-wheels : generate indexes version 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7aad7b27dc425d00416aa75b6b41891f75c35dd"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:22:04.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:29:42.000000000","tz":-180},"subject":"Removed OpenstackId legacy jobs","message":"Removed OpenstackId legacy jobs\n\nremoved legacy jobs for openstackid project and pointed\nto new migrated zuul v3 jobs\n\nDepends-On: https://review.opendev.org/702571\nNeeded-By: https://review.opendev.org/702573\nChange-Id: Id34e308509980685ed156fc68fb65975dbd0b780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff3c2157431e43bcf75815144debafbbcce9761b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff3c2157431e43bcf75815144debafbbcce9761b"}]},"branch":"refs/heads/master"},"f959cf4f075d54a28ee9118ef31b9879145d4f67":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-15 02:35:20.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/72/702572/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/702572/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/702572/3"}}},"commit":{"parents":[{"commit":"f7aad7b27dc425d00416aa75b6b41891f75c35dd","subject":"Merge \"copy-wheels : generate indexes version 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7aad7b27dc425d00416aa75b6b41891f75c35dd"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:22:04.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:34:32.000000000","tz":-180},"subject":"Removed OpenstackId legacy jobs","message":"Removed OpenstackId legacy jobs\n\nremoved legacy jobs for openstackid project and pointed\nto new migrated zuul v3 jobs\n\nChange-Id: Id34e308509980685ed156fc68fb65975dbd0b780\nDepends-On: https://review.opendev.org/702571\nNeeded-By: https://review.opendev.org/702573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f959cf4f075d54a28ee9118ef31b9879145d4f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f959cf4f075d54a28ee9118ef31b9879145d4f67"}]},"branch":"refs/heads/master"},"3d8c3354749e6f36951b3b38b3c65eebb0b3ca4a":{"kind":"REWORK","_number":4,"created":"2020-01-15 16:38:18.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/72/702572/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/702572/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/702572/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/702572/4"}}},"commit":{"parents":[{"commit":"f7aad7b27dc425d00416aa75b6b41891f75c35dd","subject":"Merge \"copy-wheels : generate indexes version 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7aad7b27dc425d00416aa75b6b41891f75c35dd"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 02:22:04.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-15 16:38:06.000000000","tz":-180},"subject":"Removed OpenstackId legacy jobs","message":"Removed OpenstackId legacy jobs\n\nremoved legacy jobs for openstackid project and pointed\nto new migrated zuul v3 jobs\n\nChange-Id: Id34e308509980685ed156fc68fb65975dbd0b780\nDepends-On: https://review.opendev.org/702571\nNeeded-By: https://review.opendev.org/702573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d8c3354749e6f36951b3b38b3c65eebb0b3ca4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d8c3354749e6f36951b3b38b3c65eebb0b3ca4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
