)]}'
{"id":"openstack%2Fproject-config~702576","triplet_id":"openstack%2Fproject-config~master~Ic1f902297d869ddafaa57f5910d99461a8a004f3","project":"openstack/project-config","branch":"master","topic":"pep503-indexes","hashtags":[],"change_id":"Ic1f902297d869ddafaa57f5910d99461a8a004f3","subject":"copy-wheels: more fixes","status":"MERGED","created":"2020-01-15 02:50:12.000000000","updated":"2020-01-15 03:34:28.000000000","submitted":"2020-01-15 03:34:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702576-1579059268418-af565c7f","meta_rev_id":"96f3aac66870f5dcb9fe47d0bbd22c0727304dcb","_number":702576,"virtual_id_number":702576,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 03:34:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-15 03:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-15 03:19:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 03:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bfca98e9a467f422a8fdbd30234e9c0e646db53","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 02:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2a0a67e21d730aec29ae5710a0884ba273b0578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:03:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3baa6e4358bd4b97a6cfe7c41b48824f : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2090e79c7f998b385d3e1fdb695c24781dfff95b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 03:19:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ni will again monitor this closely, so we don\u0027t have broken builds","accounts_in_message":[],"_revision_number":1},{"id":"34bf14beb18725dca050c251f5b754eead02609e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:19:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43571f09a828343549db8eeff434124162cb1002","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:34:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a6e49c3946ec1f34578021f898a971ac6bcaf80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:34:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f94cfbc644a4a479761c9e5bf1e45b2 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b","revisions":{"d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b":{"kind":"REWORK","_number":1,"created":"2020-01-15 02:50:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/702576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/702576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/702576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/702576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/702576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/702576/1"}}},"commit":{"parents":[{"commit":"f7aad7b27dc425d00416aa75b6b41891f75c35dd","subject":"Merge \"copy-wheels : generate indexes version 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7aad7b27dc425d00416aa75b6b41891f75c35dd"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 02:48:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 02:50:01.000000000","tz":660},"subject":"copy-wheels: more fixes","message":"copy-wheels: more fixes\n\nIterating on this further; fix incorrect yum install and match for the\ndefault Python3 install YAML.\n\nChange-Id: Ic1f902297d869ddafaa57f5910d99461a8a004f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
