)]}'
{"id":"openstack%2Fproject-config~702583","triplet_id":"openstack%2Fproject-config~master~I91f80cdccff8f1bd2e36bcd4d259f585e1baa5e1","project":"openstack/project-config","branch":"master","topic":"pep503-indexes","hashtags":[],"change_id":"I91f80cdccff8f1bd2e36bcd4d259f585e1baa5e1","subject":"copy-wheel: package install needs to be root","status":"MERGED","created":"2020-01-15 05:06:48.000000000","updated":"2020-01-15 05:24:24.000000000","submitted":"2020-01-15 05:24:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702583-1579065864173-fa858c1e","meta_rev_id":"ed3b3c5d4120dd00f03148dcad454cee76b37248","_number":702583,"virtual_id_number":702583,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 05:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-15 05:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-15 05:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 05:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2d0c6e00d3a524be0e69f5aa9265ecf03f096af","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 05:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5a3e30e5a1e9c36fdb4d21bef6603763da072a4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 05:07:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nwill monitor","accounts_in_message":[],"_revision_number":1},{"id":"45a1dd2f076b7bfc3cee8b9f1a84071380a60cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:13:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6ad242beac3428ebaf696f40572d8c6 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"08c7782c62dd95f82087cf8be22dd93909a4301d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:13:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb0248cec6687de16e01851992c4526cd4b95bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:24:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1197bcd6d776426c882cae1db3b0497d : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6a1eb3757707e6fdc37a225e2f1b979ce5dcb090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:24:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"447d64a68a710e8e1d1e9c86292d2cfee490b667","revisions":{"447d64a68a710e8e1d1e9c86292d2cfee490b667":{"kind":"REWORK","_number":1,"created":"2020-01-15 05:06:48.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/83/702583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/702583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/702583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/702583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/702583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/702583/1"}}},"commit":{"parents":[{"commit":"d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b","subject":"copy-wheels: more fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4a5fdd38ae4e5d7f498bdb62f846ab9135c038b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 05:04:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 05:06:36.000000000","tz":660},"subject":"copy-wheel: package install needs to be root","message":"copy-wheel: package install needs to be root\n\nChange-Id: I91f80cdccff8f1bd2e36bcd4d259f585e1baa5e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/447d64a68a710e8e1d1e9c86292d2cfee490b667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/447d64a68a710e8e1d1e9c86292d2cfee490b667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
