)]}'
{"id":"openstack%2Fproject-config~702673","triplet_id":"openstack%2Fproject-config~master~I8b61bdaf9a6f253dfa13fb7dd1689440a79f517a","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I8b61bdaf9a6f253dfa13fb7dd1689440a79f517a","subject":"Update compute-hyperv test jobs","status":"MERGED","created":"2020-01-15 14:40:09.000000000","updated":"2020-01-16 04:05:32.000000000","submitted":"2020-01-16 04:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702673-1579147532780-fe3cfc39","meta_rev_id":"6d802b8b7ce20e87fd4f9016a70cd094f472aa56","_number":702673,"virtual_id_number":702673,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 04:05:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-01-15 15:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-16 03:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 03:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 15:04:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-16 03:57:32.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-01-16 04:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02e67f1b2a37a54280a1c93ad055872496d310c0","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-01-15 14:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2600037175009d78de3c7b1e0b3ca4f6b178951","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 14:44:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13120531ed446a4e7b6a26598f1909921a105772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 14:53:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2251f2e06064466b8363856afb3f376f : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1f597877d526230a456c665a21aaa6623feb95ad","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-01-15 14:57:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4efe2dcc54c3b561bc1c7a5ecd572259ecffad4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 15:04:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"546868cfa64a03478c25b44dd23b960ec01a80cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:10:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26a292ab3a22443296f658039e8f1c98 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e470f18c9ef9a01f803a574a40b5557dff3a67b7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-16 03:57:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44e9eb36a4117b0a055b7f44e05cb3cd4b97d427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 03:57:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b4b28b82bf5c7c7f56d6178875666ced290a3df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:05:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6d802b8b7ce20e87fd4f9016a70cd094f472aa56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 04:05:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/62fece89d76b47af8c80cf37d378eb36 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fd8e90edf5e5447c4145d504f49aeb819a0adfe","revisions":{"689881bf5a5b616e1bb56321492be634cb0e8bcc":{"kind":"REWORK","_number":1,"created":"2020-01-15 14:40:09.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/73/702673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/702673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/702673/1"}}},"commit":{"parents":[{"commit":"3c1340bb13ae5b2b30a3eb8937e368b15ba7ecda","subject":"Merge \"copy-wheels: fix index path, remove f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c1340bb13ae5b2b30a3eb8937e368b15ba7ecda"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-01-15 14:36:11.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-01-15 14:39:54.000000000","tz":120},"subject":"Update compute-hyperv test jobs","message":"Update compute-hyperv test jobs\n\nWe\u0027re dropping Python 2.7 support, so the \"openstack-python-jobs\"\ntemplate should also be dropped. We\u0027ll do the same for the Py3.5\ntemplate and define it within the project.\n\nChange-Id: I8b61bdaf9a6f253dfa13fb7dd1689440a79f517a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/689881bf5a5b616e1bb56321492be634cb0e8bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/689881bf5a5b616e1bb56321492be634cb0e8bcc"}]},"branch":"refs/heads/master"},"8fd8e90edf5e5447c4145d504f49aeb819a0adfe":{"kind":"REWORK","_number":2,"created":"2020-01-15 14:57:49.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/73/702673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/702673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/702673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/702673/2"}}},"commit":{"parents":[{"commit":"3c1340bb13ae5b2b30a3eb8937e368b15ba7ecda","subject":"Merge \"copy-wheels: fix index path, remove f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c1340bb13ae5b2b30a3eb8937e368b15ba7ecda"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-01-15 14:36:11.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-01-15 14:56:17.000000000","tz":120},"subject":"Update compute-hyperv test jobs","message":"Update compute-hyperv test jobs\n\nWe\u0027re dropping Python 2.7 support, so the \"openstack-python-jobs\"\ntemplate should also be removed.\n\nWhile at it, we\u0027ll move all templates within the project, except for\n\"publish-to-pypi\", which is expected to stay here.\n\nChange-Id: I8b61bdaf9a6f253dfa13fb7dd1689440a79f517a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8fd8e90edf5e5447c4145d504f49aeb819a0adfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8fd8e90edf5e5447c4145d504f49aeb819a0adfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
