)]}'
{"id":"openstack%2Fproject-config~702745","triplet_id":"openstack%2Fproject-config~master~I67e67a1e98d85e5a8ed31001b26e280e29bc6318","project":"openstack/project-config","branch":"master","topic":"retire-zmq-event-publisher","hashtags":[],"change_id":"I67e67a1e98d85e5a8ed31001b26e280e29bc6318","subject":"Add noop-jobs for zmq-event-publisher","status":"MERGED","created":"2020-01-15 20:26:05.000000000","updated":"2020-01-15 21:13:19.000000000","submitted":"2020-01-15 21:13:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702745-1579122799873-17819b5d","meta_rev_id":"b9622c56660e9b015303bdbb8d28881392a970f4","_number":702745,"virtual_id_number":702745,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 21:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-15 20:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-01-15 21:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 21:02:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-15 21:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4dd8a051f426e67bf40b55c6d87f54bcae77acd7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 20:26:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9538d9033a48a8f5625f60311617e7280f015b0f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 20:26:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9d9280a05f466a479be6e1feca7d6d349ae316f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 20:29:55.000000000","message":"Topic changed from zmq-event to retire-zmq-event-publisher","accounts_in_message":[],"_revision_number":1},{"id":"591ce105a9b4e92f3cb47d03677f400fd0f927c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 20:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/33e6453009ba4e028a07340c484dce16 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4d0c285e35fa08df3369bbd7347294375f250a99","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 21:02:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbf83be0accd3980830b1caf9b2000d3bed32fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:02:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dba5cb4d00ed546698913f0bece8c09ed6033926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:13:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f022981a035647189c99226b3c55218e : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b9622c56660e9b015303bdbb8d28881392a970f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:13:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc30f930fa28d605e5bfc5ed6c7514a538bf81e2","revisions":{"fc30f930fa28d605e5bfc5ed6c7514a538bf81e2":{"kind":"REWORK","_number":1,"created":"2020-01-15 20:26:05.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/45/702745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/702745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/702745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/702745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/702745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/702745/1"}}},"commit":{"parents":[{"commit":"c7d292d803b6d7803a9b4b5c0d49873af9f58430","subject":"Revert \"Disable compression of files in stage-output role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7d292d803b6d7803a9b4b5c0d49873af9f58430"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-15 20:24:37.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-15 20:24:37.000000000","tz":60},"subject":"Add noop-jobs for zmq-event-publisher","message":"Add noop-jobs for zmq-event-publisher\n\nThis repo is not used anymore, it\u0027s been forked and taken over.\n\nRemove jobs and add noop-jobs so that we can remove jobs and retire repo\nif needed.\n\nChange-Id: I67e67a1e98d85e5a8ed31001b26e280e29bc6318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc30f930fa28d605e5bfc5ed6c7514a538bf81e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc30f930fa28d605e5bfc5ed6c7514a538bf81e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
