)]}'
{"id":"openstack%2Fproject-config~702769","triplet_id":"openstack%2Fproject-config~master~If4fd66c4c08e8b8de1877d1fbcfc6bb2b3d7dea3","project":"openstack/project-config","branch":"master","topic":"trusty-removal","hashtags":[],"change_id":"If4fd66c4c08e8b8de1877d1fbcfc6bb2b3d7dea3","subject":"Remove legacy-gearman-plugin-mavin-build-ubuntu-trusty","status":"MERGED","created":"2020-01-15 23:04:48.000000000","updated":"2020-01-16 07:41:47.000000000","submitted":"2020-01-16 07:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702769-1579160507577-a981c286","meta_rev_id":"4b6fe2dc60ab436dec6ddab75ba550e8014e357a","_number":702769,"virtual_id_number":702769,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 07:41:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-15 23:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-16 07:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-16 07:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 23:31:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-16 07:34:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-16 07:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc191f63e1a52097aae8e13c60857f6ca9424bee","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 23:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d9721d0f0394d9db934f9d8ddaf070603a9b535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 23:05:18.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"87b5d63ca153d79c12db6a30e9702aeabc47bc38","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 23:31:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7ea2192bfef7dd723eb6c362333595a7f77b713","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 07:21:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"158daed877948f82d7b9ec05b12ee329ab5910b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c9799452acc44ba59b5d500a1a2b853f : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"348611009c0def799e62ecc7182a3728992d253a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 07:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e25340cff74ac822c3fcd20a132a722ac9dadb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:35:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f3cd624d9404b4df6638795da5c7eb05a3b7711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:41:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42dbc14c068249168c605bf8848fcd0a : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b6dcbb8f2e1c12686de03fa2371fbccb2bc7a0a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:41:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c56c2d6bdf47612d11e6b75cfba2073bfaf9964","revisions":{"1c56c2d6bdf47612d11e6b75cfba2073bfaf9964":{"kind":"REWORK","_number":1,"created":"2020-01-15 23:04:48.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/69/702769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/702769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/702769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/702769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/702769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/702769/1"}}},"commit":{"parents":[{"commit":"f310e651db84801a7a804e883f65ce30409c8b0f","subject":"Remove infra trusty jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f310e651db84801a7a804e883f65ce30409c8b0f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:02:34.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:02:34.000000000","tz":660},"subject":"Remove legacy-gearman-plugin-mavin-build-ubuntu-trusty","message":"Remove legacy-gearman-plugin-mavin-build-ubuntu-trusty\n\nThe dependent change introduces a non-trusty based job, switch to that\n\nDepends-On: https://review.opendev.org/702768\nChange-Id: If4fd66c4c08e8b8de1877d1fbcfc6bb2b3d7dea3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c56c2d6bdf47612d11e6b75cfba2073bfaf9964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c56c2d6bdf47612d11e6b75cfba2073bfaf9964"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
