)]}'
{"id":"openstack%2Fproject-config~702774","triplet_id":"openstack%2Fproject-config~master~I9b272782c83cff4e456958be2eb3a38ec60aca6a","project":"openstack/project-config","branch":"master","topic":"trusty-removal","hashtags":[],"change_id":"I9b272782c83cff4e456958be2eb3a38ec60aca6a","subject":"Remove trusty wheel jobs","status":"MERGED","created":"2020-01-15 23:19:50.000000000","updated":"2020-01-18 15:25:34.000000000","submitted":"2020-01-18 15:25:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702774-1579361134854-1f963bb9","meta_rev_id":"226e50b3957a485a7bb3811fbb7890a06754068c","_number":702774,"virtual_id_number":702774,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-18 15:25:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-15 23:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-01-18 15:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-18 15:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 23:32:11.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-18 15:19:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-18 15:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca94548a4a47f89ed120cbdf8571e12ee8c5853f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 23:19:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f785f6442171ad8ef8b5ba66f183fdd4c66d5ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 23:20:21.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"fd850601f4936f40b8c573f2067e77a750b67645","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 23:32:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31f59c67c80d6573cce414379c5d040fd925eedd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 07:21:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a1e4603e6147ae7f70d91a801019b7e99c5ce5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:23:00.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch stable/queens.  The error was:\n\n  Job build-wheel-mirror-ubuntu-trusty not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-pypy-jobs-nonvoting\n        - openstack-python-jobs\n        - openstack-python35-jobs\n      check:\n        jobs:\n          - build-wheel-mirror-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-trusty:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              voting: false\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-cinder-py27\n          - cross-cinder-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-glance-py27\n          - cross-glance-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-horizon-py27\n          - cross-horizon-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-keystone-py27\n          - cross-keystone-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-neutron-py27\n          - cross-neutron-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-py27\n          - cross-nova-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-functional\n          - cross-swift-py27\n          - cross-swift-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - requirements-integration\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-neutron-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - build-wheel-mirror-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-trusty:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n          - cross-cinder-py27\n          - cross-cinder-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-glance-py27\n          - cross-glance-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-horizon-py27\n          - cross-horizon-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-keystone-py27\n          - cross-keystone-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-neutron-py27\n          - cross-neutron-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-py27\n          - cross-nova-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n              files: ^upper-constraints.txt$\n          - cross-nova-functional\n          - cross-swift-py27\n          - cross-swift-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - requirements-integration\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-neutron-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - legacy-tempest-dsvm-neutron-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@stable/queens\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c1a9c5ec14a20924971507e5b1a3f858c966749b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 16:21:52.000000000","message":"Patch Set 1:\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":1},{"id":"e230a577f5357de54bf4d2b9941050409ac9b698","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 16:22:17.000000000","message":"Patch Set 1:\n\nwe still have jobs using trusty - should we wait with merging this?","accounts_in_message":[],"_revision_number":1},{"id":"6b279974fe08061d370e7618acb94646c08dca55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:22:49.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch stable/queens.  The error was:\n\n  Job build-wheel-mirror-ubuntu-trusty not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-pypy-jobs-nonvoting\n        - openstack-python-jobs\n        - openstack-python35-jobs\n      check:\n        jobs:\n          - build-wheel-mirror-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-trusty:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              voting: false\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-cinder-py27\n          - cross-cinder-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-glance-py27\n          - cross-glance-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-horizon-py27\n          - cross-horizon-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-keystone-py27\n          - cross-keystone-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-neutron-py27\n          - cross-neutron-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-py27\n          - cross-nova-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-functional\n          - cross-swift-py27\n          - cross-swift-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - requirements-integration\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-neutron-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - build-wheel-mirror-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-trusty:\n              files:\n                - bindep.txt\n          - build-wheel-mirror-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py27-check-uc\n          - requirements-tox-py35-check-uc:\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n          - cross-cinder-py27\n          - cross-cinder-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-glance-py27\n          - cross-glance-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-horizon-py27\n          - cross-horizon-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-keystone-py27\n          - cross-keystone-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-neutron-py27\n          - cross-neutron-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - cross-nova-py27\n          - cross-nova-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n              files: ^upper-constraints.txt$\n          - cross-nova-functional\n          - cross-swift-py27\n          - cross-swift-py35:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - requirements-integration\n          - neutron-grenade-multinode:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - legacy-tempest-dsvm-neutron-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - legacy-tempest-dsvm-neutron-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@stable/queens\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a010f3223f2a41c97384b948d1c1fad16288938c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-18 06:42:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89fe9b3e1fa6d999da239170b51c01520e201275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 06:57:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6194d380e2314155b4a72ced9f119157 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":1},{"id":"968b655b8a10a08863ea320dd303ce1bafec6832","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-18 15:19:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ccf5bb0748c080688204b8fe0ce3e87a675daf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:20:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab635011765099a5b285326cdd27f0c6402182fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:25:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"815451ee1f555c404a71af81742071bf437fc42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:25:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ab4d781ebbb245e48b16da902ba3f2c0 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea","revisions":{"ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea":{"kind":"REWORK","_number":1,"created":"2020-01-15 23:19:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/74/702774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/702774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/702774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/702774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/702774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/702774/1"}}},"commit":{"parents":[{"commit":"1c56c2d6bdf47612d11e6b75cfba2073bfaf9964","subject":"Remove legacy-gearman-plugin-mavin-build-ubuntu-trusty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c56c2d6bdf47612d11e6b75cfba2073bfaf9964"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:16:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:16:53.000000000","tz":660},"subject":"Remove trusty wheel jobs","message":"Remove trusty wheel jobs\n\nRemove trusty wheel jobs in preparation for node removal.  No jobs\nshould be using the wheels any more.\n\nDepends-On: https://review.opendev.org/702773\nChange-Id: I9b272782c83cff4e456958be2eb3a38ec60aca6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
