)]}'
{"id":"openstack%2Fproject-config~703189","triplet_id":"openstack%2Fproject-config~master~I3fa4c26b0c8aeacf1af76f9046ea98edb2fcdbd0","project":"openstack/project-config","branch":"master","topic":"trusty-removal","hashtags":[],"change_id":"I3fa4c26b0c8aeacf1af76f9046ea98edb2fcdbd0","subject":"Bye, Bye, Trusty","status":"MERGED","created":"2020-01-17 20:12:16.000000000","updated":"2020-01-20 13:50:58.000000000","submitted":"2020-01-20 13:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703189-1579528258513-641ca554","meta_rev_id":"7773dfb47e05e31558ec1ec8d907daa317ddfc11","_number":703189,"virtual_id_number":703189,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 13:50:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-01-20 12:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-19 15:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-01-20 13:37:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-01-20 12:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-19 15:09:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-20 12:41:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-01-20 13:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5782be4efcd355433fb58d138328989cdcdcc15","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-17 20:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1b1ede78f801b1675fc0c21ca4bfb027e1c1b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 20:12:36.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"dac57b5f1f362c982fc118305a84ae1341820766","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-19 15:02:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8824f88246f99a7d6798c1addcfd1f3e1934de8d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-19 15:09:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb22df0482d7b8b536fe1497eca66458fd71cc52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 15:11:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/81d332c90b4043d7956130afb9019e3d : SUCCESS in 6m 17s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/70de0304103e4daba7697eeb4aa96abb : SUCCESS in 3m 04s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/29935772396b4f3ba63fc34636a8682c : SUCCESS in 2m 57s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8322f90cc9f344a1b196a9b4cd6d31b4 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"63ff040f52caefc4955a5377824c0629236f3d7f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-20 12:41:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99323e0b196115d995ea1199a2669e102881ed08","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-20 13:37:32.000000000","message":"Patch Set 2: Workflow+1\n\ndependency merged, re-approving","accounts_in_message":[],"_revision_number":2},{"id":"23b25ed18569c4f4f03457d37294ff04d85d500e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:37:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49f88c04c0aef0048ceed84d26c27a163b07b667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:50:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9631f3a05edb4c16bbc136d3468696f7 : SUCCESS in 8m 29s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/305527744c64462ca5a8119df5ff7922 : SUCCESS in 6m 14s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/3c3e31bb8bb541229dd52bded9d53580 : SUCCESS in 6m 19s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/81cd75c8561d4358a598ecbab7fef2f4 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c11aed179ce9d52d876259ac5a7b884c0bb498c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:50:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"398adb791fb3a00a0550c9822f513fba13022b47","revisions":{"ffef9c6e94e022e5a576b4ee4c08cad649b8deeb":{"kind":"REWORK","_number":1,"created":"2020-01-17 20:12:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/89/703189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/703189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/703189/1"}}},"commit":{"parents":[{"commit":"ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea","subject":"Remove trusty wheel jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 20:01:51.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 20:01:51.000000000","tz":60},"subject":"Set ubuntu-trusty in nodepool to -1","message":"Set ubuntu-trusty in nodepool to -1\n\nTo start removing ubuntu-trusty from nodepool, set min-ready to -1.\n\nDepends-On: https://review.opendev.org/702771\nDepends-On: https://review.opendev.org/702818\nChange-Id: I3fa4c26b0c8aeacf1af76f9046ea98edb2fcdbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ffef9c6e94e022e5a576b4ee4c08cad649b8deeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ffef9c6e94e022e5a576b4ee4c08cad649b8deeb"}]},"branch":"refs/heads/master"},"398adb791fb3a00a0550c9822f513fba13022b47":{"kind":"REWORK","_number":2,"created":"2020-01-19 15:02:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/89/703189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/703189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/703189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/703189/2"}}},"commit":{"parents":[{"commit":"ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea","subject":"Remove trusty wheel jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ca52e809bdfb8ec2e9433b925e232d8f2dd2a9ea"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 20:01:51.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-19 15:00:55.000000000","tz":60},"subject":"Bye, Bye, Trusty","message":"Bye, Bye, Trusty\n\nThis removes trusty from the repo and thus from OpenDev.\n\nAfterwards the AFS volume mirror.wheel.trustyx64 can be deleted.\n\nDepends-On: https://review.opendev.org/702771\nDepends-On: https://review.opendev.org/702818\nChange-Id: I3fa4c26b0c8aeacf1af76f9046ea98edb2fcdbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/398adb791fb3a00a0550c9822f513fba13022b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/398adb791fb3a00a0550c9822f513fba13022b47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
