)]}'
{"id":"openstack%2Fproject-config~703416","triplet_id":"openstack%2Fproject-config~master~Ib2a8037c93157b5d2ec7e60809451d4e8009ef08","project":"openstack/project-config","branch":"master","topic":"merges","hashtags":[],"change_id":"Ib2a8037c93157b5d2ec7e60809451d4e8009ef08","subject":"Update list for #openstack-merges","status":"ABANDONED","created":"2020-01-20 15:08:22.000000000","updated":"2020-01-22 10:03:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d992a16a0222cf9402fd7c5c4cb23db94e024633","_number":703416,"virtual_id_number":703416,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-20 15:48:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":-1,"date":"2020-01-20 17:04:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":-1,"date":"2020-01-20 15:56:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 15:48:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-20 17:04:22.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"}],"messages":[{"id":"9ef5e8df9ff6cd346bafe36c7e6b1638505a9a11","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-20 15:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1220b190928055d509e41f5afa9505b84808b222","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-20 15:21:57.000000000","message":"Patch Set 1: Workflow-1\n\nhttps://review.opendev.org/703421 is an alternative, marking this as WIP for now.","accounts_in_message":[],"_revision_number":1},{"id":"78c8a91548a3cc1a5e0b20a17a53bd8354abd101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:29:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/12912e9ac51b48d9a727c84a1d980b59 : SUCCESS in 9m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/634ac5eebed149689a724c7e26401cec : FAILURE in 11m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7fc31d6e12d0a175087fef2942d2f5b8a518cb5b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-20 15:35:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3cd4575ef95070e0431b1e355fc56dcd0c695518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:48:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/99f158b4aa4b4c5a80ef478d67fb3a36 : SUCCESS in 6m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/ef69f8413de44f09ae122e8808c868ec : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cafe3ed06361d246227788569e2b629834d9a51b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-20 15:56:48.000000000","message":"Patch Set 2: Workflow-1\n\nLet\u0027s go with the alternative - but wait a bit...","accounts_in_message":[],"_revision_number":2},{"id":"17e15d5f1de1f8758e068215b8219c3c808610e1","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-01-20 17:04:22.000000000","message":"Patch Set 2: Code-Review-1\n\nprefer the other option - no need to have a channel with one or two people in it get all of these","accounts_in_message":[],"_revision_number":2},{"id":"8275130cac5962e0b0ea6939856c8da52dab21cd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-22 10:03:52.000000000","message":"Abandoned\n\nlet\u0027s go with the other one...","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"067fedc5cc23a2303d1474a9a6c1c850438edc03","revisions":{"28440992e672997a2882ba99da8461c561bd6e36":{"kind":"REWORK","_number":1,"created":"2020-01-20 15:08:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/703416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/703416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/703416/1"}}},"commit":{"parents":[{"commit":"066b4fbf6506aacb05350136adce9c6e67c94221","subject":"Merge \"Bye, Bye, Trusty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/066b4fbf6506aacb05350136adce9c6e67c94221"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-20 15:02:34.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-20 15:07:39.000000000","tz":60},"subject":"Update list for #openstack-merges","message":"Update list for #openstack-merges\n\nUpdate the list with new repos, include the other \"official\" repos.\nNote that programs.yaml does not exist anymore.\n\nAlso, this does not include the opendev namespace anymore, I could\nchange the regex but since opendev is moving out of openstack, it is\nfine IMHO already.\n\nChange-Id: Ib2a8037c93157b5d2ec7e60809451d4e8009ef08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28440992e672997a2882ba99da8461c561bd6e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28440992e672997a2882ba99da8461c561bd6e36"}]},"branch":"refs/heads/master"},"067fedc5cc23a2303d1474a9a6c1c850438edc03":{"kind":"REWORK","_number":2,"created":"2020-01-20 15:35:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/703416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/703416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/703416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/703416/2"}}},"commit":{"parents":[{"commit":"066b4fbf6506aacb05350136adce9c6e67c94221","subject":"Merge \"Bye, Bye, Trusty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/066b4fbf6506aacb05350136adce9c6e67c94221"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-20 15:02:34.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-20 15:35:33.000000000","tz":60},"subject":"Update list for #openstack-merges","message":"Update list for #openstack-merges\n\nUpdate the list with new repos, include the other \"official\" repos.\nNote that programs.yaml does not exist anymore.\n\nAlso, this does not include the opendev namespace anymore, I could\nchange the regex but since opendev is moving out of openstack, it is\nfine IMHO already.\n\nChange-Id: Ib2a8037c93157b5d2ec7e60809451d4e8009ef08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/067fedc5cc23a2303d1474a9a6c1c850438edc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/067fedc5cc23a2303d1474a9a6c1c850438edc03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
