)]}'
{"id":"openstack%2Fproject-config~703464","triplet_id":"openstack%2Fproject-config~master~I41de5a1f57eed0a165bcd93ab1585003c3d478fb","project":"openstack/project-config","branch":"master","topic":"univeral","hashtags":[],"change_id":"I41de5a1f57eed0a165bcd93ab1585003c3d478fb","subject":"Stop forcing universal wheel releases","status":"MERGED","created":"2020-01-20 17:34:22.000000000","updated":"2020-01-20 17:57:43.000000000","submitted":"2020-01-20 17:57:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703464-1579543063826-fa858c1e","meta_rev_id":"59cb3df91dcde3a646c2b7457fe3e3ee669eaec8","_number":703464,"virtual_id_number":703464,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 17:57:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-01-20 17:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-20 17:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-20 17:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 17:37:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-20 17:41:35.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-20 17:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e5624a3209db69f33c550691abeae4f6bc2ab17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-20 17:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5562630d33da31e67a2f841b180c3084ee8ddf09","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-20 17:37:02.000000000","message":"Patch Set 1: Code-Review+2\n\nMy understanding is that for python2 consumption would be forced through the sdist instead of wheel path. Since our packages don\u0027t do linking against C libs (universal is incompatible with this btw) that should result in very minimal visible downstream impact. Users may see package installs take a little longer but otherwise continue to work.","accounts_in_message":[],"_revision_number":1},{"id":"948151387a5b76c1eed59e3125566a61c6c9bf21","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-20 17:41:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nProbably deserves an announcement to openstack-discuss with [dev][release] or similar in the subject.","accounts_in_message":[],"_revision_number":1},{"id":"e8ee5beb8d9537165be7afe678afce0994d791a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 17:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ffb9b2d66ca46abb694f34fae6d775f : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"418b1b657f8ede2bcd002320351ce7cca9fb1ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 17:46:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"661dd1f68ffadb629d60f3ab0750c719367cb473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 17:57:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eff092d10de9e7809abe737cd0cf249cac58c0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 17:57:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1849b5995e58426291727170343c4e6c : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3da4458ee785beccec65e2c018c2f5b285af2a2","revisions":{"c3da4458ee785beccec65e2c018c2f5b285af2a2":{"kind":"REWORK","_number":1,"created":"2020-01-20 17:34:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/64/703464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/703464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/703464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/703464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/703464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/703464/1"}}},"commit":{"parents":[{"commit":"1f09d5b76ffd97791fe9b794d6a16e96cd9f9249","subject":"Merge \"Add ircbot and repo for Multi-Arch SIG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1f09d5b76ffd97791fe9b794d6a16e96cd9f9249"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-20 17:31:26.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-20 17:31:26.000000000","tz":-360},"subject":"Stop forcing universal wheel releases","message":"Stop forcing universal wheel releases\n\nWe added the --universal flag to always be included in python release\njobs in I09cf4e64194395417c1bfa95237b04bd316c18fa to make sure projects\nthat did not set it in their setup.cfg would still get universal wheels.\nThis worked well while we had both py2 and py3 support, but now that we\nhave transitioned to just py3, we do not still want this to happen.\n\nWe would still like it to happen on stable releases that do still\nsupport py2, but that really is up to the project to set the flag in\nsetup.cfg. Removing this from the job definition is a trade off, but\nit\u0027s considered slightly better that we do not have universal wheels for\nour new deliverables since not having it for stable will just mean\nslightly longer install times.\n\nChange-Id: I41de5a1f57eed0a165bcd93ab1585003c3d478fb\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3da4458ee785beccec65e2c018c2f5b285af2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3da4458ee785beccec65e2c018c2f5b285af2a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
