)]}'
{"id":"openstack%2Fproject-config~703691","triplet_id":"openstack%2Fproject-config~master~Id7f80fef33c3b2f97345e21187250dd2e7a28ba9","project":"openstack/project-config","branch":"master","topic":"charm-interface-keystone-notifications","hashtags":[],"change_id":"Id7f80fef33c3b2f97345e21187250dd2e7a28ba9","subject":"Add charm-interface-keystone-notifications project","status":"MERGED","created":"2020-01-21 20:43:13.000000000","updated":"2020-01-24 09:34:12.000000000","submitted":"2020-01-24 09:34:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703691-1579858452110-17819b5d","meta_rev_id":"8729e4c72956a4be035c1994461d69970114d2ba","_number":703691,"virtual_id_number":703691,"owner":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 09:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-01-23 15:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-01-24 09:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-01-23 14:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-01-24 09:21:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 14:55:35.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2020-01-23 15:54:26.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-24 09:21:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-01-24 09:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71941adac04370bf787f2e754faf4ea2a48d083e","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2020-01-21 20:43:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf42a03e2a9e653d007eb3908b62e79e89adbcc","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2020-01-21 20:44:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1508c284d3b111cee0c3438e2ae4fba5bd9297dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:54:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3466c20df82c4312a93b5cd369b6126a : SUCCESS in 8m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/38aa863bb8ee4150a0fd7d91823edaa6 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f815bfb40c8fc8e0a7b4b90a4289fb12aa49bb4a","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2020-01-23 14:40:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19047c17411952172ebe5c28ac88bf447b15d6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 14:55:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f1abedb2c5854449b38322773eaf4820 : SUCCESS in 9m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c2eff5f43d2848f6a3aad0d5d5f7efe4 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9c7c223a1c8a5182cdf64056d1a4064fe7e87cde","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-01-23 14:55:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5d389a85f35decbb6db07abc3362b1eaca3a507c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-23 15:54:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"acc5eb6edbd0a241c6a09f27bcf4819281384476","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-24 09:21:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0961bfcf72e74ffaf40da37731703b8b0be7ef6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:21:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"89681a72bcd8e9bfedbb3a986ac97448a0a6b1d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:34:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d649f2445da48269b3fcf1ae9e2c6d7 : SUCCESS in 9m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4a760f6e1c714d369efc336b8669506c : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ad2ad2cb6a572f268f123cd914b0fdfc2870d3b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:34:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"704ec6c66d3fddc0c891738af6b3bf98406fb35b","revisions":{"1f9782c4b3581576d4936f1ca87566df337e7f9b":{"kind":"REWORK","_number":1,"created":"2020-01-21 20:43:13.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/91/703691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/703691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/703691/1"}}},"commit":{"parents":[{"commit":"703f1077885158e4dbda7f97a5f012448e84c80c","subject":"Fix use of \u0027comment-added\u0027 event type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/703f1077885158e4dbda7f97a5f012448e84c80c"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-21 20:22:35.000000000","tz":-180},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-21 20:26:05.000000000","tz":-180},"subject":"Add charm-interface-keystone-notifications project","message":"Add charm-interface-keystone-notifications project\n\nThis patch adds charm-interface-keytone-notifications project to enable\nthe use of identity-notifications relation in reactive charms.\n\nChange-Id: Id7f80fef33c3b2f97345e21187250dd2e7a28ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1f9782c4b3581576d4936f1ca87566df337e7f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1f9782c4b3581576d4936f1ca87566df337e7f9b"}]},"branch":"refs/heads/master"},"bbf7010b9050598d726917e7dde428c92758fff0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-21 20:44:24.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/91/703691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/703691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/703691/2"}}},"commit":{"parents":[{"commit":"703f1077885158e4dbda7f97a5f012448e84c80c","subject":"Fix use of \u0027comment-added\u0027 event type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/703f1077885158e4dbda7f97a5f012448e84c80c"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-21 20:22:35.000000000","tz":-180},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-21 20:43:57.000000000","tz":-180},"subject":"Add charm-interface-keystone-notifications project","message":"Add charm-interface-keystone-notifications project\n\nThis patch adds charm-interface-keytone-notifications project to enable\nthe use of identity-notifications relation in reactive charms.\n\nNeeded-By: https://review.opendev.org/703692\nChange-Id: Id7f80fef33c3b2f97345e21187250dd2e7a28ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbf7010b9050598d726917e7dde428c92758fff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbf7010b9050598d726917e7dde428c92758fff0"}]},"branch":"refs/heads/master"},"704ec6c66d3fddc0c891738af6b3bf98406fb35b":{"kind":"REWORK","_number":3,"created":"2020-01-23 14:40:01.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/91/703691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/703691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/703691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/703691/3"}}},"commit":{"parents":[{"commit":"703f1077885158e4dbda7f97a5f012448e84c80c","subject":"Fix use of \u0027comment-added\u0027 event type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/703f1077885158e4dbda7f97a5f012448e84c80c"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-21 20:22:35.000000000","tz":-180},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2020-01-23 14:39:29.000000000","tz":-180},"subject":"Add charm-interface-keystone-notifications project","message":"Add charm-interface-keystone-notifications project\n\nThis patch adds charm-interface-keytone-notifications project to enable\nthe use of identity-notifications relation in reactive charms.\n\nNeeded-By: https://review.opendev.org/703692\nChange-Id: Id7f80fef33c3b2f97345e21187250dd2e7a28ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704ec6c66d3fddc0c891738af6b3bf98406fb35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704ec6c66d3fddc0c891738af6b3bf98406fb35b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
