)]}'
{"id":"openstack%2Fproject-config~703917","triplet_id":"openstack%2Fproject-config~master~Ib6797c5bcc90d492c4a47e173b93b3867548879f","project":"openstack/project-config","branch":"master","topic":"linaro-us","hashtags":[],"change_id":"Ib6797c5bcc90d492c4a47e173b93b3867548879f","subject":"Add linaro-us cloud to nodepool","status":"MERGED","created":"2020-01-23 05:18:42.000000000","updated":"2020-01-24 09:27:50.000000000","submitted":"2020-01-24 09:27:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703917-1579858070527-17819b5d","meta_rev_id":"84f8da5b9486b5d1f1ea26db7a2ecdbb0e0fd717","_number":703917,"virtual_id_number":703917,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 09:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 22:17:21.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-01-23 06:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-01-24 09:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-01-24 09:20:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 06:57:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-24 09:20:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-01-24 09:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e42a09e8d581ce71b1adcc336cd07ba25a06201","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-23 05:18:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9877286fd478aa28b67b48c591d3ceaa5c56619c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 05:26:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/063c8b8fc4f141b594c1665ae7564fbb : SUCCESS in 6m 54s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e5bf913e02e6498d82cb1a5ff4e22053 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"233761d158fb460fd5f91d35824fb9d3bb3937d3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-23 05:43:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c160b5a36cbf3f28b4f7e0f281362e5b850433de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 05:51:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4f94a7fdbe534f97947b9fc7490221cb : SUCCESS in 7m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/032fa2f1338f46a19df125be0f7a8ab6 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"52e68908162af0d9286aa6f8bebf8e49173a0cfd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-23 06:57:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fb22c65c88db8c9fa589c7e299000678ef31d22f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-23 09:24:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd813ba1dfecbe3521e1af48a390d90124c313eb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-23 22:17:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a694cf3764b1afd8169402550c0d9341459b2159","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-24 09:20:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1e2111c4d26d52e15ea4f7ba4639d34f05faf0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:20:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c7dc02259a9f91e25047092bfd05ae83a6be30d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:27:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"230d2a1e20541929db953ca413fe307f8123b137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:27:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ad5f09bb00f42cabdf1e06fa626f594 : SUCCESS in 6m 31s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/420d313204144f719ed63061326f383f : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b56b83840ec4e05637bb3947ed03b2e536bf60d8","revisions":{"93bf2aac8f7e231b6c878e843111e37b1432f044":{"kind":"REWORK","_number":1,"created":"2020-01-23 05:18:42.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/17/703917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/703917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/703917/1"}}},"commit":{"parents":[{"commit":"261f8b1e79cd6e683e4788b616bc33c847faf543","subject":"Correct LOGS path in wheel build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/261f8b1e79cd6e683e4788b616bc33c847faf543"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-23 05:16:27.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-23 05:18:00.000000000","tz":660},"subject":"Add linaro-us cloud to nodepool","message":"Add linaro-us cloud to nodepool\n\nWe have enough quota for 44 8-node hosts; give ourselves a little\nheadroom.  The mirror host is up and responding.\n\nStory: #2007195\nTask: #38357\nChange-Id: Ib6797c5bcc90d492c4a47e173b93b3867548879f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93bf2aac8f7e231b6c878e843111e37b1432f044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93bf2aac8f7e231b6c878e843111e37b1432f044"}]},"branch":"refs/heads/master"},"b56b83840ec4e05637bb3947ed03b2e536bf60d8":{"kind":"REWORK","_number":2,"created":"2020-01-23 05:43:55.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/17/703917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/703917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/703917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/703917/2"}}},"commit":{"parents":[{"commit":"261f8b1e79cd6e683e4788b616bc33c847faf543","subject":"Correct LOGS path in wheel build script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/261f8b1e79cd6e683e4788b616bc33c847faf543"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-23 05:16:27.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-23 05:43:37.000000000","tz":660},"subject":"Add linaro-us cloud to nodepool","message":"Add linaro-us cloud to nodepool\n\nWe have enough quota for 44 8-node hosts; give ourselves a little\nheadroom.  The mirror host is up and responding.\n\nStory: #2007195\nTask: #38357\nChange-Id: Ib6797c5bcc90d492c4a47e173b93b3867548879f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b56b83840ec4e05637bb3947ed03b2e536bf60d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b56b83840ec4e05637bb3947ed03b2e536bf60d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
