)]}'
{"id":"openstack%2Fproject-config~704306","triplet_id":"openstack%2Fproject-config~master~I2d744c1d4dfe13a795c0d6754520d1bbf39162b5","project":"openstack/project-config","branch":"master","topic":"check-release-approval","hashtags":[],"change_id":"I2d744c1d4dfe13a795c0d6754520d1bbf39162b5","subject":"check-release-approval job improvements","status":"MERGED","created":"2020-01-27 11:10:14.000000000","updated":"2020-01-28 08:39:51.000000000","submitted":"2020-01-28 08:39:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704306-1580200791844-728c2603","meta_rev_id":"9fa9a5c18d7da250a14dd8c5fee79b2a199070bf","_number":704306,"virtual_id_number":704306,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 08:39:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-01-28 08:28:19.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-01-27 13:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-01-28 08:28:19.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 13:54:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-28 08:28:19.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-28 08:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f3024c37715a33ed868805f161cab5a94e69351","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 11:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a8720ce404a0f0e93f8fca18cf3e54b1a36c90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 11:10:33.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected str @ data[\u0027allowed-projects\u0027][0]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: check-release-approval\n      description: |\n        Checks that release was approved by PTL or release liaison.\n      files:\n        - ^deliverables/.*$\n      allowed-projects:\n        - name: openstack/releases\n      required-projects:\n        - name: openstack/governance\n      run: playbooks/check-release-approval/run.yaml\n      final: true\n      timeout: 120\n      nodeset:\n        nodes: []\n\n  in \"openstack/project-config/zuul.d/jobs.yaml@master\", line 1130, column 3","accounts_in_message":[],"_revision_number":1},{"id":"60fcb057cd4fb03cc2b45e65b70349080a92b256","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-27 11:58:12.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ae7edec4747ed7e76d45152ea4699f8ddde72f29","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 12:29:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd1c5224ba3ae24d6d9600da7c2fafe2fe2e9459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 12:37:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/997a68f7de0f4c6c843a272e206f2081 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c708d69c76f13f29b5273764d515704d703850b0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-27 13:54:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f01ca3837ed79374c62084e92855fa4fbcbf23f4","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-28 08:28:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"089a8066486f45687cc9e0639bb7a3cf713794d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:28:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dafb83ade8070c3f02063d9d320dc05be94d9d5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:39:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c26c0aa9ca89953d807ecd1622dc0154f01175d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:39:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5a196b29983247979eb0c8c0b1ffd994 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9bce2b3de88c3cb6547454fa55cb0a02b65d4d65","revisions":{"bdf849d1116f34cda5b0114904491fa151cfc070":{"kind":"REWORK","_number":1,"created":"2020-01-27 11:10:14.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/06/704306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/704306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/704306/1"}}},"commit":{"parents":[{"commit":"a4e3094c8594c3ddaeab90396984421351163ffd","subject":"Adapt branches for sdk projects in gerritbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4e3094c8594c3ddaeab90396984421351163ffd"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 11:04:56.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 11:09:23.000000000","tz":60},"subject":"check-release-approval job improvements","message":"check-release-approval job improvements\n\nLimit check-release-approval to openstack/releases since no\nother repository should be using it.\n\nAlso fix a corner case where release requests combined with\nother changes would crash the check code (only changes in\ndeliverable files should be taken into account).\n\nChange-Id: I2d744c1d4dfe13a795c0d6754520d1bbf39162b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdf849d1116f34cda5b0114904491fa151cfc070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdf849d1116f34cda5b0114904491fa151cfc070"}]},"branch":"refs/heads/master"},"9bce2b3de88c3cb6547454fa55cb0a02b65d4d65":{"kind":"REWORK","_number":2,"created":"2020-01-27 12:29:33.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/06/704306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/704306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/704306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/704306/2"}}},"commit":{"parents":[{"commit":"a4e3094c8594c3ddaeab90396984421351163ffd","subject":"Adapt branches for sdk projects in gerritbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4e3094c8594c3ddaeab90396984421351163ffd"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 11:04:56.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 12:29:08.000000000","tz":60},"subject":"check-release-approval job improvements","message":"check-release-approval job improvements\n\nLimit check-release-approval to openstack/releases since no\nother repository should be using it.\n\nAlso fix a corner case where release requests combined with\nother changes would crash the check code (only changes in\ndeliverable files should be taken into account).\n\nChange-Id: I2d744c1d4dfe13a795c0d6754520d1bbf39162b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bce2b3de88c3cb6547454fa55cb0a02b65d4d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bce2b3de88c3cb6547454fa55cb0a02b65d4d65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
