)]}'
{"id":"openstack%2Fproject-config~704411","triplet_id":"openstack%2Fproject-config~master~Ie22b9dfe7b8840e6e8b84b85dd4d2d0101ee9b7b","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"Ie22b9dfe7b8840e6e8b84b85dd4d2d0101ee9b7b","subject":"Add opendev/pre-commit-hooks","status":"ABANDONED","created":"2020-01-27 20:07:41.000000000","updated":"2020-04-15 14:48:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"013b1e6c4dba13fb52dc3d62fca457b20a43322e","_number":704411,"virtual_id_number":704411,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-27 12:33:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 20:08:42.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2020-03-25 18:47:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-27 12:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8b649e01e4d0dc88f7dcd7095c1c564bda6b1db","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-01-27 20:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"479c18577a7602266a30dcc34643c803f5d9e36d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 20:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5abc351451e34808a9324459d468fe73 : SUCCESS in 7m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/46cf9b132b154eea8eacc5dd5df529f4 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"69f576258711603f56d35ee62cbc27852b00b077","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-25 18:47:46.000000000","message":"Patch Set 1:\n\nare you still interested in this and will fix it - or time to abandon?","accounts_in_message":[],"_revision_number":1},{"id":"16f4876704c65282fc5f871680bcaa02e3a466da","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-27 12:01:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0260a13386aa29c9544c1fadc159f51517be51ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 12:12:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/abc042713ca142dd98f1b602b7e3bfe9 : SUCCESS in 8m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e01839da14914fab8c6948155d0a00e7 : FAILURE in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"59ae61616638f5e6f0b238d1c0bcc1bf9443e7c6","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-27 12:21:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"475c51a8eba6060f9eb3a0e5d3fdfe2354f101b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 12:33:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5c242a5140314de5996d87dd66fc66a0 : SUCCESS in 7m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/000c9722ced143c89db0b8ca67069d31 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"03ebcd42b9dbc7692fe8d4146ca29972fa977892","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-04-15 14:48:03.000000000","message":"Abandoned\n\nAbandoning because I am not working on it, but happy to reopen if somoene wants to pursuit it.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbd3984b36eec097d9ca6b2d66b63dc7953c77f2","revisions":{"b036f93f4591d8ebcf4c525028ddb7f9442c6a06":{"kind":"REWORK","_number":1,"created":"2020-01-27 20:07:41.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/11/704411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/704411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/704411/1"}}},"commit":{"parents":[{"commit":"a4e3094c8594c3ddaeab90396984421351163ffd","subject":"Adapt branches for sdk projects in gerritbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4e3094c8594c3ddaeab90396984421351163ffd"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-01-27 20:01:02.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-01-27 20:01:02.000000000","tz":0},"subject":"Add opendev/pre-commit-hooks","message":"Add opendev/pre-commit-hooks\n\nThis should allow use to easily define re-usable hooks to be shared\nacross our projects instead of relying on upstream version from github.\n\nChange-Id: Ie22b9dfe7b8840e6e8b84b85dd4d2d0101ee9b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b036f93f4591d8ebcf4c525028ddb7f9442c6a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b036f93f4591d8ebcf4c525028ddb7f9442c6a06"}]},"branch":"refs/heads/master"},"fbd3984b36eec097d9ca6b2d66b63dc7953c77f2":{"kind":"REWORK","_number":2,"created":"2020-03-27 12:21:50.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/11/704411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/704411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/704411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/704411/2"}}},"commit":{"parents":[{"commit":"a4e3094c8594c3ddaeab90396984421351163ffd","subject":"Adapt branches for sdk projects in gerritbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4e3094c8594c3ddaeab90396984421351163ffd"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-01-27 20:01:02.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-03-27 12:21:24.000000000","tz":0},"subject":"Add opendev/pre-commit-hooks","message":"Add opendev/pre-commit-hooks\n\nThis should allow use to easily define re-usable hooks to be shared\nacross our projects instead of relying on upstream version from github.\n\nChange-Id: Ie22b9dfe7b8840e6e8b84b85dd4d2d0101ee9b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbd3984b36eec097d9ca6b2d66b63dc7953c77f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbd3984b36eec097d9ca6b2d66b63dc7953c77f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
