)]}'
{"id":"openstack%2Fproject-config~704648","triplet_id":"openstack%2Fproject-config~master~I56a9bfc7bb7bd0180bcbab842514c77e9081e54e","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I56a9bfc7bb7bd0180bcbab842514c77e9081e54e","subject":"Don\u0027t require CLA for contributors of Ansible Openstack modules","status":"MERGED","created":"2020-01-28 18:07:08.000000000","updated":"2020-01-29 18:01:06.000000000","submitted":"2020-01-29 18:01:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704648-1580320866930-9ea6cd4b","meta_rev_id":"3aa214513e56c4b22293415b0ea3fad7ba773a47","_number":704648,"virtual_id_number":704648,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 18:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-01-29 11:09:32.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-01-29 17:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-01-29 17:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 18:14:14.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2020-01-28 18:14:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-01-28 18:14:30.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-01-29 06:42:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-29 11:09:32.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-29 17:46:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-29 18:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"905d304c8d43a192c96d77aaedf784b3fd04ebb1","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-01-28 18:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b410f2a032810cbddbc00093b7e29558d76a54ca","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-01-28 18:08:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5b25e4d321829c4cd964a7509d804640f66251b2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 18:12:10.000000000","message":"Patch Set 1:\n\nThis seems like a reasonable choice to me. For new contributions happening under a CLA to mean anything, all former contributors would probably also need to be contacted and asked to agree to it as well.","accounts_in_message":[],"_revision_number":1},{"id":"94952022ef7922c19205a9b5b4777948bd577e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:18:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/81f9698f775a4635b74bac0f5e63a84f : SUCCESS in 8m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d76637efd3e84916b74f9ef71c46d7f9 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9ff7d10a8d1890ebfc606b824c170e1de69f0c93","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-29 04:48:17.000000000","message":"Patch Set 1: Code-Review+2\n\nI do not think the CLA is necessary for this repo. The repo is not an OpenStack deliverable - it is a repo owned by a sig. This choice was made purposely because the code is from ansible and is GPL\u0027d.","accounts_in_message":[],"_revision_number":1},{"id":"4f7565a236938201547136f01824f3816fe048ed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-29 06:42:46.000000000","message":"Patch Set 1: Workflow-1\n\nRemove the WIP from the title and we can +2A ;)","accounts_in_message":[],"_revision_number":1},{"id":"e6effae0a1502a21060d842f309675eb3b79018c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-01-29 09:55:32.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"803800a74ac1280e5f834b4d4cd95bad51fa7b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 10:05:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be2a9f56c4914caab165b7d76d77980e : SUCCESS in 7m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/93cbf3bf37dd49ea9532b48e980bb0e4 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"15f95091df4c1b24ee7a17ec70c4a4c46592198e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-29 11:09:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd06160377107ce85d0af1128a4497b92f92af98","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-29 17:46:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nAlso, as Monty points out, these are GPL sources so the OSF CLA is entirely inapplicable as it\u0027s Apache License specific.","accounts_in_message":[],"_revision_number":2},{"id":"eb2a148a1c7ea4f6724d04a0ab566ed3d2c92828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:47:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb1b8e9b12a79ab95c1b900834423f271fb8cb56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 18:01:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3aa214513e56c4b22293415b0ea3fad7ba773a47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 18:01:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/820cf15dbf744819b460692cd9a43abc : SUCCESS in 8m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/379c3e5d6a7a47ecaa8f0eb662e1ae95 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65bf2b3faafe00b76f8094d27fbcb664e139029a","revisions":{"2cabdb46515d91f793cf9b10bbf68cb144fd7931":{"kind":"REWORK","_number":1,"created":"2020-01-28 18:07:08.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/48/704648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/704648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/704648/1"}}},"commit":{"parents":[{"commit":"adc64344ddb9a9f68dad21b523bd3d36bd4a80ab","subject":"Merge \"Updated x/networking-mlnx project for pypi and neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/adc64344ddb9a9f68dad21b523bd3d36bd4a80ab"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-01-28 18:05:29.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-01-28 18:07:05.000000000","tz":120},"subject":"WIP: don\u0027t require CLA for contributors of Ansible Openstack modules","message":"WIP: don\u0027t require CLA for contributors of Ansible Openstack modules\n\nHistory:\n  - Ansible Openstack modules are moved from Ansible core repo to\n    Openstack SIG repo. Like all other modules will be separated\n    from Ansible core in 2.10 and will go somwhere.\n  - Ansible contributors are confused by moving to Gerrit and\n    because of requirement to sign for CLA.\n\nAs it\u0027s SIG repo and not part of \"official\" Openstack, can we SKIP\nrequirement of signing of CLA for contributors?\n\nChange-Id: I56a9bfc7bb7bd0180bcbab842514c77e9081e54e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2cabdb46515d91f793cf9b10bbf68cb144fd7931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2cabdb46515d91f793cf9b10bbf68cb144fd7931"}]},"branch":"refs/heads/master"},"65bf2b3faafe00b76f8094d27fbcb664e139029a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-29 09:55:32.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/48/704648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/704648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/704648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/704648/2"}}},"commit":{"parents":[{"commit":"adc64344ddb9a9f68dad21b523bd3d36bd4a80ab","subject":"Merge \"Updated x/networking-mlnx project for pypi and neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/adc64344ddb9a9f68dad21b523bd3d36bd4a80ab"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-01-28 18:05:29.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-01-29 09:55:29.000000000","tz":0},"subject":"Don\u0027t require CLA for contributors of Ansible Openstack modules","message":"Don\u0027t require CLA for contributors of Ansible Openstack modules\n\nHistory:\n  - Ansible Openstack modules are moved from Ansible core repo to\n    Openstack SIG repo. Like all other modules will be separated\n    from Ansible core in 2.10 and will go somwhere.\n  - Ansible contributors are confused by moving to Gerrit and\n    because of requirement to sign for CLA.\n\nAs it\u0027s SIG repo and not part of \"official\" Openstack, can we SKIP\nrequirement of signing of CLA for contributors?\n\nChange-Id: I56a9bfc7bb7bd0180bcbab842514c77e9081e54e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65bf2b3faafe00b76f8094d27fbcb664e139029a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65bf2b3faafe00b76f8094d27fbcb664e139029a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
