)]}'
{"id":"openstack%2Fproject-config~705547","triplet_id":"openstack%2Fproject-config~master~I64fc41d9caaef6cb3fc06df0a5f712d7e2940063","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I64fc41d9caaef6cb3fc06df0a5f712d7e2940063","subject":"Add \"Review-Priority\" gerrit label to devstack","status":"MERGED","created":"2020-02-03 20:23:07.000000000","updated":"2020-02-05 06:19:09.000000000","submitted":"2020-02-05 06:10:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705547-1580883017203-bbef5f2b","meta_rev_id":"3e2c881866e95bb32986f1849f40755d51bc6247","_number":705547,"virtual_id_number":705547,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-02-04 18:48:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 06:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-02-05 02:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-02-04 23:02:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-05 05:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"date":"2020-02-04 19:20:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-05 05:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 19:20:49.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-04 23:02:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-05 02:57:41.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-05 05:57:48.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-02-05 06:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08d59d59f5a74ba02a4c88d4708b136f77e7057b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-03 20:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a246a7d17e2931ed57490804feb9bb5c43bb1990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 20:32:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9cde2cae917546ad93b5e050424c30c7 : SUCCESS in 7m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e6c3f229ddea4449864cf2831bf556b3 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"11f09ec9d7fbd5b229c58be4cee6262c5759cc2d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-03 21:08:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8468e801b26d3eff17a93e75415967f9fe4918b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 07:23:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59ad76da85c850ac93e0e980331a41af69f5465a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 07:24:18.000000000","message":"Patch Set 2:\n\nyeah, I feel bad for forgetting to save my edit yesterday :-)","accounts_in_message":[],"_revision_number":2},{"id":"9f86dfa81d1754e0ec92834448a1bd72e36ac5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 07:34:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/01b6f5c5a97a4086a0c068315c95404e : SUCCESS in 7m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/baa56f6f99054c9685fff63179a04b2d : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"811f42757edcc7d9a50821a95df98f77e3ab51a4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 18:08:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0ba0aafb778a57cc1779f240d74937654a6ee6ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 18:36:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a83493d9beb2e0eef79a673d4c4056155e3b78b9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 18:36:59.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"afe136b5e2a4ebe9c1e70b382bdc48a81bdb64a2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 18:44:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a14833fc03c0136686fa0f6db5dd5595f8b00a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 18:46:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/770267231aa947e8bb466621c26fea52 : SUCCESS in 8m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d13230a8e63495ea275ca9b293125ca : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4},{"id":"370c9b5ff0d30d1e5e60522633722ea4f3b81ec7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 18:46:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8c6956a80eea2154cdecc60a25bdfd8b5ca7670","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 18:48:12.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"578326664607ee54813a0ab9fa4aa729202279eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-04 18:48:23.000000000","message":"Patch Set 5:\n\nhacking gerrit at its finest :-)","accounts_in_message":[],"_revision_number":5},{"id":"7202c1aa20b8aa38c18823affc9c069fcfe2171e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 19:00:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e7dbf914fb9a456da18c86dc1913f43a : SUCCESS in 8m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a819d04b344d4826aba4188552f39235 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":5},{"id":"6aefcf0f6f454225f62c258f0340d8e7699106d4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 23:02:53.000000000","message":"Patch Set 5: Code-Review+1\n\nthanks. lgtm.","accounts_in_message":[],"_revision_number":5},{"id":"0679220ca460d2353e13c3e6c493604129238bc2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-05 02:57:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b9de4b7b2dba62fc222a07c8f72889612375924b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-05 05:57:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9e59ca338e43c61af5d7d4d5ddd35cdfd6422d8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 05:58:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5802bc13de3974f617e8009b61f8b1c547758c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 06:10:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c546405f6ec4f158ac976e3abffdb6e : SUCCESS in 7m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/840c0d9945334306bbb2474b58ea0ef1 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"957ef18802f79ef50ef50b95c5b938ddc43ab514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 06:10:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3e2c881866e95bb32986f1849f40755d51bc6247","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 06:19:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/5f7dd87007eb4ebe9e4c39be1461b49b : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5111b88fdb2a2d52378dc1886fae31bb804fa5aa","revisions":{"70620ab42a003a6a81822c66f86ef25243f4d806":{"kind":"REWORK","_number":1,"created":"2020-02-03 20:23:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/705547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/705547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/705547/1"}}},"commit":{"parents":[{"commit":"a1155eb355e0b0d13e7b64e0505501f447902b2d","subject":"Merge \"Artifact to AFS publishing - don\u0027t use archive flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1155eb355e0b0d13e7b64e0505501f447902b2d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"subject":"Add RP to devstack","message":"Add RP to devstack\n\nChange-Id: I64fc41d9caaef6cb3fc06df0a5f712d7e2940063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70620ab42a003a6a81822c66f86ef25243f4d806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70620ab42a003a6a81822c66f86ef25243f4d806"}]},"branch":"refs/heads/master"},"323025bd952fad4e65b56db9fa97c66e2fbff215":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-04 07:23:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/705547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/705547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/705547/2"}}},"commit":{"parents":[{"commit":"a1155eb355e0b0d13e7b64e0505501f447902b2d","subject":"Merge \"Artifact to AFS publishing - don\u0027t use archive flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1155eb355e0b0d13e7b64e0505501f447902b2d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-04 07:23:48.000000000","tz":0},"subject":"Add \"Review-Priority\" gerrit label to devstack","message":"Add \"Review-Priority\" gerrit label to devstack\n\nThis is to align devstack with many other OpenStack projects.\n\nChange-Id: I64fc41d9caaef6cb3fc06df0a5f712d7e2940063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/323025bd952fad4e65b56db9fa97c66e2fbff215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/323025bd952fad4e65b56db9fa97c66e2fbff215"}]},"branch":"refs/heads/master"},"790e56cc42061de7c3c202e00e7b24f80cbe653a":{"kind":"REWORK","_number":3,"created":"2020-02-04 18:36:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/705547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/705547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/705547/3"}}},"commit":{"parents":[{"commit":"a1155eb355e0b0d13e7b64e0505501f447902b2d","subject":"Merge \"Artifact to AFS publishing - don\u0027t use archive flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1155eb355e0b0d13e7b64e0505501f447902b2d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-04 18:36:07.000000000","tz":0},"subject":"Add \"Review-Priority\" gerrit label to devstack","message":"Add \"Review-Priority\" gerrit label to devstack\n\nThis is to align devstack with many other OpenStack projects.\n\nChange-Id: I64fc41d9caaef6cb3fc06df0a5f712d7e2940063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/790e56cc42061de7c3c202e00e7b24f80cbe653a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/790e56cc42061de7c3c202e00e7b24f80cbe653a"}]},"branch":"refs/heads/master"},"04cf178cc2cbeecb00a50d8cfa99d9ea9a41e94b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-04 18:36:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/705547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/705547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/705547/4"}}},"commit":{"parents":[{"commit":"a1155eb355e0b0d13e7b64e0505501f447902b2d","subject":"Merge \"Artifact to AFS publishing - don\u0027t use archive flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1155eb355e0b0d13e7b64e0505501f447902b2d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-04 18:36:56.000000000","tz":0},"subject":"Add \"Review-Priority\" gerrit label to devstack","message":"Add \"Review-Priority\" gerrit label to devstack\n\nThis is to align devstack with many other OpenStack projects.\nAdapted to devstack (no -1 for branch freeze nor +1 for\na regular important change).\n\nChange-Id: I64fc41d9caaef6cb3fc06df0a5f712d7e2940063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04cf178cc2cbeecb00a50d8cfa99d9ea9a41e94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04cf178cc2cbeecb00a50d8cfa99d9ea9a41e94b"}]},"branch":"refs/heads/master"},"5111b88fdb2a2d52378dc1886fae31bb804fa5aa":{"kind":"REWORK","_number":5,"created":"2020-02-04 18:48:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/705547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/705547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/705547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/705547/5"}}},"commit":{"parents":[{"commit":"a1155eb355e0b0d13e7b64e0505501f447902b2d","subject":"Merge \"Artifact to AFS publishing - don\u0027t use archive flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1155eb355e0b0d13e7b64e0505501f447902b2d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-03 20:22:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-04 18:48:09.000000000","tz":0},"subject":"Add \"Review-Priority\" gerrit label to devstack","message":"Add \"Review-Priority\" gerrit label to devstack\n\nThis is to align devstack with many other OpenStack projects.\nAdapted to devstack (no -1 for branch freeze nor +1 for\na regular important change).\n\nChange-Id: I64fc41d9caaef6cb3fc06df0a5f712d7e2940063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5111b88fdb2a2d52378dc1886fae31bb804fa5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5111b88fdb2a2d52378dc1886fae31bb804fa5aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
