)]}'
{"id":"openstack%2Fproject-config~705990","triplet_id":"openstack%2Fproject-config~master~Ieab04a4d6c02b216a59c12ec8599e7d91f4fffb1","project":"openstack/project-config","branch":"master","topic":"check-approval-pipeline","hashtags":[],"change_id":"Ieab04a4d6c02b216a59c12ec8599e7d91f4fffb1","subject":"Add release-approval pipeline","status":"MERGED","created":"2020-02-05 15:50:53.000000000","updated":"2020-02-05 20:44:06.000000000","submitted":"2020-02-05 20:39:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705990-1580935151102-5b512e01","meta_rev_id":"8f3caa46b5a7f2467de3e45d1f616460839210be","_number":705990,"virtual_id_number":705990,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 20:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-02-05 15:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-02-05 16:03:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-05 20:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-02-05 20:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 15:57:29.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-02-05 16:03:30.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-05 20:28:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-05 20:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"103626f166ece1ec5bf5621846931ac18fdc441d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-05 15:50:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6948cff7fd57e90f849c7c8ec1c34329d0cc3886","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-02-05 15:57:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e668776291569bd88d0fc2effd8879ffde0bfb2","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-05 16:03:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2662bc5191dd8085cd101c5bec8651792f8ff8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:03:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f108750712964e95b2bf838ffa160193 : SUCCESS in 7m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eaa64c1748684b248044e6040c4d2ed1 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"73286175e177b63e25a742818b02a520199bba87","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-05 20:28:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"556c50763926edba15e66a14f91b8f9fe3ad517e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:28:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17ac9d46f65da1e4e5a132c92d0c7530c6d7efd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:39:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e02194c7e4b044d1963813566076582f : SUCCESS in 8m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7743cbf4c0c74eaa9dee5d162bf5ac9c : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ec4318cc91534201ab78cc8539fbdb581ce5546e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:39:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"997d4ee2c80e9eec676f48b993bed1ff81b478aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:44:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/42b54697458546a9be3c47c206101789 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e896b57eb6e6f1be97df0841c897765d8c1f11d","revisions":{"7e896b57eb6e6f1be97df0841c897765d8c1f11d":{"kind":"REWORK","_number":1,"created":"2020-02-05 15:50:53.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/90/705990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/705990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/705990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/705990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/705990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/705990/1"}}},"commit":{"parents":[{"commit":"2a6570399c8562ff830886d6716b1a5f6fb9a187","subject":"Merge \"Add \"Review-Priority\" gerrit label to devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a6570399c8562ff830886d6716b1a5f6fb9a187"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-05 15:25:29.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-05 15:46:24.000000000","tz":60},"subject":"Add release-approval pipeline","message":"Add release-approval pipeline\n\nDefine a release-approval pipeline to run the check-release-approval\njob on every comment added to a release request, and set a\nPTL-Approved label accordingly.\n\nThis may be considered a bit resource-intensive, however the\ncheck-release-approval job is a fast python script that runs on\nthe executor, and only release requests shall go in this pipeline.\nIf this generates too much load, we could configure it to only run\nwhen the comment posted contains a magic \"signoff\" keyword.\n\nAnother concern is that jobs other than check-release-approval would\nbe added to this pipeline. There does not seem to be a way in Zuul to\nlimit a pipeline to a specific job name or project.\n\nChange-Id: Ieab04a4d6c02b216a59c12ec8599e7d91f4fffb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e896b57eb6e6f1be97df0841c897765d8c1f11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e896b57eb6e6f1be97df0841c897765d8c1f11d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
