)]}'
{"id":"openstack%2Fproject-config~706186","triplet_id":"openstack%2Fproject-config~master~Idfc286489bdc9d89e2bae9316e31607fca244a54","project":"openstack/project-config","branch":"master","topic":"remove-unnecessary-job-from-networking-spp","hashtags":[],"change_id":"Idfc286489bdc9d89e2bae9316e31607fca244a54","subject":"Remove unnecessary jobs from networking-spp","status":"MERGED","created":"2020-02-06 06:57:15.000000000","updated":"2020-02-06 08:36:50.000000000","submitted":"2020-02-06 07:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706186-1580974242649-1e8c484d","meta_rev_id":"36b70fb4d6e862d7f92a7dc3b0b0bfe048a998f6","_number":706186,"virtual_id_number":706186,"owner":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 07:30:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-02-06 07:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"value":2,"date":"2020-02-06 07:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"value":1,"date":"2020-02-06 07:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 07:02:08.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-02-06 07:13:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-06 07:21:00.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-02-06 07:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bae1324eb14fa2e31e16ea56121f68d8a73061f4","author":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"date":"2020-02-06 06:57:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff22651967eb41488487cdc18f8ae6e999687aff","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 07:02:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb560069347a7deba2501756052889d2acddf522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:08:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cb6887dbb2a848fb9725b53c73ce782c : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9b25287a33e1ee66da4e162e75581d1fb478abbb","author":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"date":"2020-02-06 07:10:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4be250e8107da97fd4fefd7cc68ae5ad8601a4f3","author":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"date":"2020-02-06 07:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello Andreas,\n\nThank you for reviewing. I fixed this patch according to your comment.\n\nThe following patch adds check-requirements to networking-spp\u0027s zuul.yaml.\n\nhttps://review.opendev.org/#/c/705895/","accounts_in_message":[],"_revision_number":1},{"id":"441bfb6ad1e9598b1d15e3c78306b73af6dfa7c7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 07:13:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ce34492855adbeb0a53d4db7d90e22cd2faa461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:18:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fb95d774608f46c6b850d3a9c766d775 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a6b4d4b1171cd16c9c049dc0d3c03a430c17af65","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-06 07:21:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"12b6468235631e1150579ce1d95a7f756e663d5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:21:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a57c4e947d1bcf2a9debe728f42745ba52f5435","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b90822b18f96149b4d3f8aa9468deba130df8012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:30:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0a80d58586d74f6b8d869619360ee27f : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"65854497ed5ec50da103acef0f19dfd8bb301c0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:35:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/de6d8630c4f84d30bf750c8a1c1da187 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"df22a7c64f659c8a69df283f49342ab3e4c7be9e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 08:33:50.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nLet me revert","accounts_in_message":[],"_revision_number":2},{"id":"9c987d984525e8bc6de90f0d6e0c506b4235d4a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 08:34:14.000000000","message":"Created a revert of this change as Id9b2bb34145cb513b49885d208a29ea6d439e0ab","accounts_in_message":[],"_revision_number":2},{"id":"6fea4df49965d53b42e91aa5e3e308f9b55071b5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 08:36:50.000000000","message":"Patch Set 2:\n\nSorry, let me abandon revert - this might be right and I got confused by the vpp change in flight as well.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04d0df0a44e4855cba9f48de2be214ab781a670b","revisions":{"074e7bd57fde7533cb010d81b60aaead4581cbe3":{"kind":"REWORK","_number":1,"created":"2020-02-06 06:57:15.000000000","uploader":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"ref":"refs/changes/86/706186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/706186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/706186/1"}}},"commit":{"parents":[{"commit":"2a6570399c8562ff830886d6716b1a5f6fb9a187","subject":"Merge \"Add \"Review-Priority\" gerrit label to devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a6570399c8562ff830886d6716b1a5f6fb9a187"}]}],"author":{"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","date":"2020-02-06 06:45:19.000000000","tz":0},"committer":{"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","date":"2020-02-06 06:56:50.000000000","tz":0},"subject":"Remove openstack-python-jobs-neutron job from networking-spp","message":"Remove openstack-python-jobs-neutron job from networking-spp\n\nPython 2.7 support and tests were already dropped from networking-spp.\nAnd openstack-python-jobs-neutron job contains openstack-tox-py27 job.\nSo this patch removes openstack-python-jobs-neutron job from\nnetworking-spp.\nOther tests contained in openstack-python-jobs-neutron are contained\nin jobs which specified in networking-spp\u0027s zuul.yaml.\n\nChange-Id: Idfc286489bdc9d89e2bae9316e31607fca244a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074e7bd57fde7533cb010d81b60aaead4581cbe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/074e7bd57fde7533cb010d81b60aaead4581cbe3"}]},"branch":"refs/heads/master"},"04d0df0a44e4855cba9f48de2be214ab781a670b":{"kind":"REWORK","_number":2,"created":"2020-02-06 07:10:22.000000000","uploader":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"ref":"refs/changes/86/706186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/706186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/706186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/706186/2"}}},"commit":{"parents":[{"commit":"2a6570399c8562ff830886d6716b1a5f6fb9a187","subject":"Merge \"Add \"Review-Priority\" gerrit label to devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a6570399c8562ff830886d6716b1a5f6fb9a187"}]}],"author":{"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","date":"2020-02-06 06:45:19.000000000","tz":0},"committer":{"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","date":"2020-02-06 07:09:38.000000000","tz":0},"subject":"Remove unnecessary jobs from networking-spp","message":"Remove unnecessary jobs from networking-spp\n\nPython 2.7 support and tests were already dropped from networking-spp.\nAnd openstack-python-jobs-neutron job contains openstack-tox-py27 job.\nSo this patch removes openstack-python-jobs-neutron job from\nnetworking-spp.\nOther tests contained in openstack-python-jobs-neutron are contained\nin jobs which specified in networking-spp\u0027s zuul.yaml.\n\nAnd, this patch also removes check-requirements job from\nnetworking-spp.\ncheck-requirements job is specified in networking-spp\u0027s zuul.yaml.\n\nChange-Id: Idfc286489bdc9d89e2bae9316e31607fca244a54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04d0df0a44e4855cba9f48de2be214ab781a670b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04d0df0a44e4855cba9f48de2be214ab781a670b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
