)]}'
{"id":"openstack%2Fproject-config~706316","triplet_id":"openstack%2Fproject-config~master~Ie0293d0523f8a9a738015035725ea2e40b152d3d","project":"openstack/project-config","branch":"master","topic":"airship-ci-cloud","hashtags":[],"change_id":"Ie0293d0523f8a9a738015035725ea2e40b152d3d","subject":"Add capacity to airship CI cloud","status":"MERGED","created":"2020-02-06 16:36:15.000000000","updated":"2020-02-07 21:54:07.000000000","submitted":"2020-02-07 21:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706316-1581112306388-18a4250f","meta_rev_id":"57ebf77e300a2978327a149bc3468d865bc575f2","_number":706316,"virtual_id_number":706316,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 21:51:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"all":[{"value":1,"date":"2020-02-07 09:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-02-07 21:41:01.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-06 17:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"date":"2020-02-07 19:42:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-02-07 21:41:01.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 17:25:32.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-07 09:00:27.000000000","updated_by":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"reviewer":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"state":"REVIEWER"},{"updated":"2020-02-07 21:41:01.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-07 21:51:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d5a914bdf9519848c8179fc2b3a87fe0ceaf46f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-06 16:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a428d4410897fe14d6f77f5365312bc7ccaebe41","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-06 16:36:38.000000000","message":"Patch Set 1: Workflow-1\n\nWIP until mirror is booted and quotas updated to reflect this.","accounts_in_message":[],"_revision_number":1},{"id":"86ad918d0b08023fffff2cdf6a57f5a1459e387a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 16:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ba0a5b7dc1674e529c9ec2cedd31c799 : SUCCESS in 8m 39s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/704b5a916bb340129fc9bfee9fedaeb8 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e87a885b38e49b3b3086bccefdd4427f221deef2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 17:25:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5dd5e74677dd3c689ddbdf044460b610922dc149","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-02-07 09:00:27.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. Thank you, Clark.","accounts_in_message":[],"_revision_number":1},{"id":"cada027021918fb2b4675f3520639298541662a9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-07 19:42:26.000000000","message":"Patch Set 1: -Workflow\n\nThe mirror is up and quotas are set. I think we can land this now.","accounts_in_message":[],"_revision_number":1},{"id":"a2f16d1cb4c06731d19f1062ab933423af1e0fe8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-07 21:41:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31e8d6f554a1184952fcda805b94090347f0d8c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:41:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4258e4627980d27ef143cb7ee888a99e196d95a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:51:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18d54bafeac51438367f93aa720ce581505928d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:51:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3b8ec0b4b8b34885b11784be2f808fc6 : SUCCESS in 9m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/00118077e843473c9a80b72030b58186 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5a173f3ce94f395d092e28c4a3f9a309e96e1891","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/7dd3bbbd4df441308fcd4328c963e990 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"282fbcb0d27b5c290c12809de69a65068c96180c","revisions":{"282fbcb0d27b5c290c12809de69a65068c96180c":{"kind":"REWORK","_number":1,"created":"2020-02-06 16:36:15.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/16/706316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/706316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/706316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/706316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/706316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/706316/1"}}},"commit":{"parents":[{"commit":"7d6c978e70b2ca0a015dc4a144f3046942fbfba8","subject":"Fix triggers for release-approval pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d6c978e70b2ca0a015dc4a144f3046942fbfba8"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-02-06 16:34:52.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-02-06 16:34:52.000000000","tz":-480},"subject":"Add capacity to airship CI cloud","message":"Add capacity to airship CI cloud\n\nThe airship jobs are expected to need 1x32GB ram instance + 3x16GB ram\ninstances each. The goal is to run 4 jobs concurrently which gives us 16\nmax-instances in the airship pool. We set the main pool to max 10\ninstances of the smaller 8GB ram instances to act as canaries in the\nmain pool.\n\nChange-Id: Ie0293d0523f8a9a738015035725ea2e40b152d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/282fbcb0d27b5c290c12809de69a65068c96180c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/282fbcb0d27b5c290c12809de69a65068c96180c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
