)]}'
{"id":"openstack%2Fproject-config~706319","triplet_id":"openstack%2Fproject-config~master~I22bc23b923199052bdb1f916885b8195dbe94d3a","project":"openstack/project-config","branch":"master","topic":"ideas","hashtags":[],"change_id":"I22bc23b923199052bdb1f916885b8195dbe94d3a","subject":"Add OpenStack ideas promote jobs","status":"MERGED","created":"2020-02-06 16:50:37.000000000","updated":"2020-02-17 15:18:58.000000000","submitted":"2020-02-17 15:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706319-1581952738532-3390009c","meta_rev_id":"e941e878c5770443d8f89779a7165c7c39681700","_number":706319,"virtual_id_number":706319,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2020-02-10 08:48:47.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 15:18:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-02-17 15:08:46.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-11 19:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-02-11 10:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-02-17 15:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-02-17 15:08:46.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 10:03:04.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-11 19:57:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-17 15:05:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-02-17 15:08:46.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-17 15:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de9509e38cfe438e35725ffd67df24c151604405","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-06 16:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a16269116cc8c09e54908c3aa01bb94263059d8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-06 16:56:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d08fa4707c1f73f86e7b197b5541fc30dda0671d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 17:08:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8bc54c29a5c9492ca8448f4981a35dd6 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e831ac53d31589d3d79413eb71af16f51ced7d3f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 17:37:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11b2ed1aadb09bcd01b969f291e0e6d3be870b2f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-07 10:38:34.000000000","message":"Patch Set 2:\n\nQuestion: by making this governance.o.o/ideas (instead of governance.o.o/tc/ideas), it\u0027s a bit unclear where it falls governance-wise.\n\nOther entries under governance.o.o are all governance bodies (Board, TC, UC), or constructs that derive from governance bodies (SIGs) or directly related to their lifecycle (election).\n\nWho tracks/drives the Ideas ? If it\u0027s the TC I feel like this should live under /tc/ideas. If it\u0027s a separate group of people, I feel like it should live under /sigs/ideas (probably with a Ideas-Meta SIG to drive it). If it\u0027s more of a self-sustaining dashboard with no central authority, then yes it makes sense under /ideas.\n\nWhich one is it?","accounts_in_message":[],"_revision_number":2},{"id":"9264c6a72bc6e6a951abd4f27e71eedff7a90f30","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-07 10:44:54.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"df11f27f779b0ba8bb92456d7ab82e8f700f5ed7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-10 08:47:00.000000000","message":"Patch Set 2:\n\n\u003e Question: by making this governance.o.o/ideas (instead of\n \u003e governance.o.o/tc/ideas), it\u0027s a bit unclear where it falls\n \u003e governance-wise.\n \u003e \n \u003e Other entries under governance.o.o are all governance bodies\n \u003e (Board, TC, UC), or constructs that derive from governance bodies\n \u003e (SIGs) or directly related to their lifecycle (election).\n \u003e \n \u003e Who tracks/drives the Ideas ? If it\u0027s the TC I feel like this\n \u003e should live under /tc/ideas. If it\u0027s a separate group of people, I\n \u003e feel like it should live under /sigs/ideas (probably with a\n \u003e Ideas-Meta SIG to drive it). If it\u0027s more of a self-sustaining\n \u003e dashboard with no central authority, then yes it makes sense under\n \u003e /ideas.\n \u003e \n \u003e Which one is it?\n\nSelf sustained with no central authority. That\u0027s the model.","accounts_in_message":[],"_revision_number":2},{"id":"daca9ca798abae231106ef01c25ccb4eea5c0a53","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-10 08:47:32.000000000","message":"Patch Set 2:\n\nAnswered there.","accounts_in_message":[],"_revision_number":2},{"id":"b4c6c7ed5eb1754800b9234e193980f996851316","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-10 08:47:48.000000000","message":"Patch Set 2:\n\nAnswered there.","accounts_in_message":[],"_revision_number":2},{"id":"0469a8be6565226e4b4a2f160cc679be74dd5dab","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-10 08:48:47.000000000","message":"Patch Set 2:\n\n\u003e Answered there.\n\nWow, wrong clicking. Damn mornings...","accounts_in_message":[],"_revision_number":2},{"id":"dcc8bd82069e4e3bbd05148f9d0c96f8c79aa6b7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-10 09:53:30.000000000","message":"Patch Set 2:\n\nHow does self-sustain work in practice ? Who approves changes to the openstack/ideas repository ? (Is everything accepted after a cursory review?)","accounts_in_message":[],"_revision_number":2},{"id":"9b32255ced269ba3d37e739d1a04454208d4ef92","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-11 10:03:04.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003cevrardjp\u003e ttx: about your questions on ideas, indeed the intention was that anyone could merge","accounts_in_message":[],"_revision_number":2},{"id":"f78720cec65327041f3af487095d591d2a794312","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-11 19:57:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c3b78b016a802f58712cb59d3f74a8325c694406","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-17 15:05:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e215d53685be11457baa72cc4750c0745f96f662","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-17 15:08:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6183f1ff40054441d5d7a5ead29ca84ff47c68eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:09:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc746cfe80b75e30ab0c23ba050061e24adb0817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:18:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f2f8a29d1e2a4be4adf46560e1d2a0ce : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7def015a4e047a6dbe28f107f80a7068ca95e599","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:18:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c801f1124fc90bcdf8c2ae7603c0e9d984d54e17","revisions":{"6f5493110713359a448fc87e93f656103b5a9371":{"kind":"REWORK","_number":1,"created":"2020-02-06 16:50:37.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/706319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/706319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/706319/1"}}},"commit":{"parents":[{"commit":"7d6c978e70b2ca0a015dc4a144f3046942fbfba8","subject":"Fix triggers for release-approval pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d6c978e70b2ca0a015dc4a144f3046942fbfba8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-02-06 16:30:32.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-02-06 16:35:38.000000000","tz":60},"subject":"Add promote jobs","message":"Add promote jobs\n\nThis will publish the built sphinx webpages of the ideas.\n\nNeeded-By: https://review.opendev.org/#/c/706240\nChange-Id: I22bc23b923199052bdb1f916885b8195dbe94d3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f5493110713359a448fc87e93f656103b5a9371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f5493110713359a448fc87e93f656103b5a9371"}]},"branch":"refs/heads/master"},"c801f1124fc90bcdf8c2ae7603c0e9d984d54e17":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-06 16:56:48.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/706319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/706319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/706319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/706319/2"}}},"commit":{"parents":[{"commit":"7d6c978e70b2ca0a015dc4a144f3046942fbfba8","subject":"Fix triggers for release-approval pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d6c978e70b2ca0a015dc4a144f3046942fbfba8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-02-06 16:30:32.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-02-06 16:56:45.000000000","tz":0},"subject":"Add OpenStack ideas promote jobs","message":"Add OpenStack ideas promote jobs\n\nThis will make possible to upload the files of the ideas repo into\nthe right folder for publication onto governance.o.o/ideas.\n\nNeeded-By: https://review.opendev.org/#/c/706240\nChange-Id: I22bc23b923199052bdb1f916885b8195dbe94d3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c801f1124fc90bcdf8c2ae7603c0e9d984d54e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c801f1124fc90bcdf8c2ae7603c0e9d984d54e17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
