)]}'
{"id":"openstack%2Fproject-config~706479","triplet_id":"openstack%2Fproject-config~master~Ib47416608d6bfe0f0179bbd201d90ada86463e5f","project":"openstack/project-config","branch":"master","topic":"gov-sync-cleanup","hashtags":[],"change_id":"Ib47416608d6bfe0f0179bbd201d90ada86463e5f","subject":"Fix ACLs for deliverables no longer under RelMgt","status":"MERGED","created":"2020-02-07 12:08:53.000000000","updated":"2020-02-17 15:03:20.000000000","submitted":"2020-02-17 15:03:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706479-1581951800858-a5ae4dff","meta_rev_id":"dcd595de31f30c1b58499ee48e5f3172653ba5c5","_number":706479,"virtual_id_number":706479,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 15:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-07 13:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-17 14:48:39.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-17 09:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-17 14:48:39.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 13:25:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-17 09:16:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-17 14:48:39.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-17 15:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0043ba65b463486973a5909522d8a563b43d0358","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-07 12:08:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"437672a5c9ac709e6f4704474a5bf309c2e90a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:16:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7bf87df2bda745709c0293d658c5cf94 : SUCCESS in 6m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0c17f0c3fce54e0a980f778c36684a4b : FAILURE in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3b79de151eabf318c400744e9ba85f266d7a46d2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-07 12:23:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c626332357c45fe2369b62fe71964724df2865aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:34:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2e91b3dfc7e647f5b319029be8b890db : SUCCESS in 6m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d3118c422edd4b369e8eb3f22b455e98 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2d1e2c9c9474e269f2614434712bb3aced0357cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-07 13:25:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2317fb201473c21c2acaf1bb70aca6bb9cec0369","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-17 09:16:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e080dc3d70ec4e44d23837eb5b1e11040f0d87f2","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-17 14:48:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ebd968ed818ed433ee2e965a7987e46844a8c3b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:48:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f8facf3d945006afc7a098dc0140aab43620ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:03:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6aa7efb2a2e14308bce50690410c0a75 : SUCCESS in 7m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d1f349c054bb4028870e7aeac2394bf0 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d1fa62a05b46538a0f531ab61f64cdfe706e3ffe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:03:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28a83141a0e55f67e41366c4c31f3b135b116972","revisions":{"5830df461c1ad07622f425be680e7497e1af75a2":{"kind":"REWORK","_number":1,"created":"2020-02-07 12:08:53.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/79/706479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/706479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/706479/1"}}},"commit":{"parents":[{"commit":"23d169ebf940cfa1951d28da80e7e0e37ee41f6f","subject":"stop generating constraints for py3.4 and py3.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23d169ebf940cfa1951d28da80e7e0e37ee41f6f"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 11:59:50.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 12:06:24.000000000","tz":60},"subject":"Fix ACLs for deliverables no longer under RelMgt","message":"Fix ACLs for deliverables no longer under RelMgt\n\nsyntribos (Security SIG) and python-tempest-conf (Interop WG) are\nnot handled by the release management team (which handles OpenStack\nrelease deliverables only).\n\nnova-powervm, networking-powervm and ceilometer-powervm are deliverables\nfrom the new PowerVMStackers SIG, and should also no longer be under\ncentralized release management from now on.\n\nAdjust ACLs to allow those projects to be tagged directly.\n\nChange-Id: Ib47416608d6bfe0f0179bbd201d90ada86463e5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5830df461c1ad07622f425be680e7497e1af75a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5830df461c1ad07622f425be680e7497e1af75a2"}]},"branch":"refs/heads/master"},"28a83141a0e55f67e41366c4c31f3b135b116972":{"kind":"REWORK","_number":2,"created":"2020-02-07 12:23:12.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/79/706479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/706479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/706479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/706479/2"}}},"commit":{"parents":[{"commit":"23d169ebf940cfa1951d28da80e7e0e37ee41f6f","subject":"stop generating constraints for py3.4 and py3.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23d169ebf940cfa1951d28da80e7e0e37ee41f6f"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 11:59:50.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 12:23:03.000000000","tz":60},"subject":"Fix ACLs for deliverables no longer under RelMgt","message":"Fix ACLs for deliverables no longer under RelMgt\n\nsyntribos (Security SIG) and python-tempest-conf (Interop WG) are\nnot handled by the release management team (which handles OpenStack\nrelease deliverables only).\n\nnova-powervm, networking-powervm and ceilometer-powervm are deliverables\nfrom the new PowerVMStackers SIG, and should also no longer be under\ncentralized release management from now on.\n\nAdjust ACLs to allow those projects to be tagged directly.\n\nChange-Id: Ib47416608d6bfe0f0179bbd201d90ada86463e5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28a83141a0e55f67e41366c4c31f3b135b116972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28a83141a0e55f67e41366c4c31f3b135b116972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
