)]}'
{"id":"openstack%2Fproject-config~706488","triplet_id":"openstack%2Fproject-config~master~I3e7aba14360e4df2e4d7c8f386d3fb9c8b92fa73","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I3e7aba14360e4df2e4d7c8f386d3fb9c8b92fa73","subject":"Add Label definition for PTL-Approved","status":"MERGED","created":"2020-02-07 12:36:05.000000000","updated":"2020-02-07 13:07:18.000000000","submitted":"2020-02-07 12:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706488-1581080377825-159bbb0d","meta_rev_id":"992dbc83bb0fa342d20a4f6a8b5acc9bc05219df","_number":706488,"virtual_id_number":706488,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 12:59:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-02-07 12:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-07 12:42:27.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-02-07 12:42:27.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 12:39:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-07 12:42:27.000000000","updated_by":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"reviewer":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"state":"REVIEWER"},{"updated":"2020-02-07 12:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17e7fc7848b8fd510138793e02be9a0efc2f369a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-07 12:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"616e911a5e6205c2b15589a545fd20cd4ded61ae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-07 12:39:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, I expect that our linter will not be happy.","accounts_in_message":[],"_revision_number":1},{"id":"f1ea6df21acd3d95bbec15aeaf3625e260f484b3","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-07 12:39:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81a3d57ffce7acf4005ccbda5d0d6f1434fa2aab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-07 12:39:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e3c1fc20e4abf65f5b023c7e361072fe84383fed","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2020-02-07 12:42:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af8f7cd036dfd8b6a62bdfb6870178d21975fdcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/355114d2206e440da3a9731dd306a754 : SUCCESS in 7m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/89b3e08975004f1481fbb28ca4975166 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f2d81a358c8a2bdae861c7ed384f595657055ba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:52:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18ccc81fc561a574039c31834205914902ccf22a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:59:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b8e110a692294d89b16e30f045bf050e : SUCCESS in 6m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6d6723e698e0499683f6410b9dc7a732 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"fc0997b3d5b9553c78ce1800660548b7f2ec2911","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"992dbc83bb0fa342d20a4f6a8b5acc9bc05219df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:07:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/4a3251240b9d461a9211bc23668737c2 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83bebd1d1bd8122eb6171209c4ed6eda8e6ce4be","revisions":{"b0c0c4bda81fbd0d4efcfd28f53e1715e09b152e":{"kind":"REWORK","_number":1,"created":"2020-02-07 12:36:05.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/88/706488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/706488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/706488/1"}}},"commit":{"parents":[{"commit":"23d169ebf940cfa1951d28da80e7e0e37ee41f6f","subject":"stop generating constraints for py3.4 and py3.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23d169ebf940cfa1951d28da80e7e0e37ee41f6f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-07 12:35:15.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-07 12:35:15.000000000","tz":-360},"subject":"Add Label definition for PTL-Approved","message":"Add Label definition for PTL-Approved\n\nCurrently we\u0027re just seeing \"Allow\" in:\n\n  https://review.opendev.org/#/admin/projects/openstack/releases,access\n\nMaybe it\u0027s because there is no definition of the label qualities.\n\nChange-Id: I3e7aba14360e4df2e4d7c8f386d3fb9c8b92fa73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b0c0c4bda81fbd0d4efcfd28f53e1715e09b152e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b0c0c4bda81fbd0d4efcfd28f53e1715e09b152e"}]},"branch":"refs/heads/master"},"83bebd1d1bd8122eb6171209c4ed6eda8e6ce4be":{"kind":"REWORK","_number":2,"created":"2020-02-07 12:39:41.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/88/706488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/706488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/706488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/706488/2"}}},"commit":{"parents":[{"commit":"23d169ebf940cfa1951d28da80e7e0e37ee41f6f","subject":"stop generating constraints for py3.4 and py3.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/23d169ebf940cfa1951d28da80e7e0e37ee41f6f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-07 12:35:15.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-07 12:39:38.000000000","tz":-360},"subject":"Add Label definition for PTL-Approved","message":"Add Label definition for PTL-Approved\n\nCurrently we\u0027re just seeing \"Allow\" in:\n\n  https://review.opendev.org/#/admin/projects/openstack/releases,access\n\nMaybe it\u0027s because there is no definition of the label qualities.\n\nChange-Id: I3e7aba14360e4df2e4d7c8f386d3fb9c8b92fa73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83bebd1d1bd8122eb6171209c4ed6eda8e6ce4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83bebd1d1bd8122eb6171209c4ed6eda8e6ce4be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
