)]}'
{"id":"openstack%2Fproject-config~706790","triplet_id":"openstack%2Fproject-config~master~I0d9c01324202270da9826d5490609288b11901ad","project":"openstack/project-config","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I0d9c01324202270da9826d5490609288b11901ad","subject":"Remove networking-vpp entry","status":"MERGED","created":"2020-02-10 10:46:55.000000000","updated":"2020-02-12 18:48:49.000000000","submitted":"2020-02-12 18:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706790-1581532914985-25e6796c","meta_rev_id":"6319aece61a4d29615149d54d7584c5570e5e1ab","_number":706790,"virtual_id_number":706790,"owner":{"_account_id":26839,"name":"Onong Tayeng","email":"onong.tayeng@gmail.com","username":"onong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 18:41:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":2,"date":"2020-02-12 18:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-02-10 10:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-02-11 17:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"date":"2020-02-10 10:49:12.000000000","_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"value":1,"date":"2020-02-12 18:25:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-02-11 16:30:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 10:49:12.000000000","updated_by":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"reviewer":{"_account_id":3217,"name":"Ian Wells","username":"ijw-ubuntu"},"state":"REVIEWER"},{"updated":"2020-02-10 10:49:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-11 16:30:00.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-02-11 17:52:49.000000000","updated_by":{"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},"reviewer":{"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},"state":"REVIEWER"},{"updated":"2020-02-12 18:25:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-12 18:41:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b964ebff60e65a09a566fcee115b134408513c6b","author":{"_account_id":26839,"name":"Onong Tayeng","email":"onong.tayeng@gmail.com","username":"onong"},"date":"2020-02-10 10:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99a5a41fdb6c4f66b6ee41571c9d727557b39515","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-10 10:49:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"094e7188ee5f13db9d1772e2b18ff9b18a8737e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:03:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee8a0e27614f4ae783b6349205b03923 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0cc78fd6b82f3df73c29442f23c5acfab92afe2e","author":{"_account_id":26839,"name":"Onong Tayeng","email":"onong.tayeng@gmail.com","username":"onong"},"date":"2020-02-10 13:14:00.000000000","message":"Topic set to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"1996fb2221838e539b6b53b97685d1d0ea7d3faa","author":{"_account_id":1356,"name":"Naveen Joy","email":"najoy@cisco.com","username":"najoy"},"date":"2020-02-11 17:52:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ac89ece005848ba5afa9c9ee94997bb847e77c8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-12 18:25:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d41765f9cdd79c93dbbd4d11acf40b10a240673f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:25:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5a60f5240480adf96c11ed0f9b4401ffd9c06cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:41:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4f9ad02736b492b2b514bf78e165d49cfd3facd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:41:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25cce570ee3a465ab5daec07f269f72d : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6319aece61a4d29615149d54d7584c5570e5e1ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:48:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/7600bc2d736549919a19917a68f70bac : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"731d0caf6fe95ebc1d0cd09424def0e89e443068","revisions":{"731d0caf6fe95ebc1d0cd09424def0e89e443068":{"kind":"REWORK","_number":1,"created":"2020-02-10 10:46:55.000000000","uploader":{"_account_id":26839,"name":"Onong Tayeng","email":"onong.tayeng@gmail.com","username":"onong"},"ref":"refs/changes/90/706790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/706790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/706790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/706790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/706790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/706790/1"}}},"commit":{"parents":[{"commit":"a56b4c1c82788bf32abc892fdc8a8269d840aef9","subject":"Set airship-citycloud mirror url to opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a56b4c1c82788bf32abc892fdc8a8269d840aef9"}]}],"author":{"name":"Onong Tayeng","email":"otayeng@cisco.com","date":"2020-02-10 10:38:21.000000000","tz":330},"committer":{"name":"Onong Tayeng","email":"otayeng@cisco.com","date":"2020-02-10 10:38:53.000000000","tz":330},"subject":"Remove networking-vpp entry","message":"Remove networking-vpp entry\n\nAs part of dropping py27 support, openstack-python3-ussuri-jobs\nhas been added to networking-vpp\u0027s .zuul.yaml file:\n\nhttps://review.opendev.org/#/c/705394/\n\nHence, removing the networking-vpp entry in zuul.d/projects.yaml\nas it is no longer needed anymore.\n\nChange-Id: I0d9c01324202270da9826d5490609288b11901ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/731d0caf6fe95ebc1d0cd09424def0e89e443068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/731d0caf6fe95ebc1d0cd09424def0e89e443068"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
