)]}'
{"id":"openstack%2Fproject-config~707086","triplet_id":"openstack%2Fproject-config~master~Ia13b651ce99ca295775187dc653e246b84fd1f43","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ia13b651ce99ca295775187dc653e246b84fd1f43","subject":"End gating for networking-calico, as retiring project from OpenStack","status":"MERGED","created":"2020-02-11 09:41:52.000000000","updated":"2020-02-17 07:18:40.000000000","submitted":"2020-02-17 07:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707086-1581923429310-bbef5f2b","meta_rev_id":"91187263d00e55438eafec23a719d4ea2840f91d","_number":707086,"virtual_id_number":707086,"owner":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-02-13 23:33:29.000000000","_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 07:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-02-12 10:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-17 06:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-17 06:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 10:57:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-17 06:57:02.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-02-17 07:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9620a60afd26630aea330ae715f957c9bd8d41f1","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2020-02-11 09:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03308e8d89d7e34fdaf8f839b637b42af3956b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 09:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a9860ea78fcc43b1bb8e8a88fc49f879 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1effb47f9134688a5fd9445efd628fe598bef422","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-11 19:58:03.000000000","message":"Patch Set 1:\n\nPlease push the followup changes as well, I\u0027ll +2 once I\u0027ve seen the rest...","accounts_in_message":[],"_revision_number":1},{"id":"0c19bfb6ff54d0ac1c504bfb9fc0e96e99cf3198","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2020-02-11 22:01:28.000000000","message":"Patch Set 1:\n\nAndreas, can you clarify what you mean?  Are there more changes needed in this repo?\n\n(I noticed there are remaining occurrences of \"networking-calico\", but they aren\u0027t mentioned in Step 2, so I guessed they would eventually be cleaned up via some other process.)","accounts_in_message":[],"_revision_number":1},{"id":"abb468d4233102b36cda78a7d838ae647b381604","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2020-02-11 22:02:06.000000000","message":"Patch Set 1:\n\nSorry, to be clear, I\u0027m referring to Step 2 of https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project","accounts_in_message":[],"_revision_number":1},{"id":"f23f39674527e53d468ee2e91beecba2c0cf5cb5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-12 05:30:27.000000000","message":"Patch Set 1:\n\nI\u0027d like to see changes for the other steps from the infra manual.\n\nThis is step 2 of 4 according to https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nSo, please push steps 3 and 4 as well.","accounts_in_message":[],"_revision_number":1},{"id":"95a7dc066641780ead6cf9486dc411cf42b268c3","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2020-02-12 10:49:58.000000000","message":"Patch Set 1:\n\nThanks Andreas, I\u0027ve prepared those next changes now:\n\nStep 3: https://review.opendev.org/#/c/707357/\nStep 4: https://review.opendev.org/#/c/707360/","accounts_in_message":[],"_revision_number":1},{"id":"101b6bd0d5f431b0fa4d2f509be79141e624994d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-12 10:57:48.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"cdce1d82be0af3eba7c4132a2544d109175db6fd","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2020-02-13 23:33:29.000000000","message":"Patch Set 1:\n\nHi Andreas - I just thought I should check: is this just waiting now for one of your colleagues to review and grant another +2, or is there something else that *I* should be doing to help this along?","accounts_in_message":[],"_revision_number":1},{"id":"4d537780d39483b7a96328a6bc5984add43bada5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-14 05:48:07.000000000","message":"Patch Set 1:\n\nYes, this waits for a second reviewer.","accounts_in_message":[],"_revision_number":1},{"id":"2ecc59652a45274fab0238b238e23fe08ea95482","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-17 06:57:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb218bb5b2dcc22755d0ee5d51f649f1e4ed99eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 06:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50c546e874b4041fc45936431f780eb4b7add521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 07:10:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a04c58fb613d71b8bb21f831529f763b2dd97bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 07:10:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9898a23f7fee434f9b804548c8802dad : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"91187263d00e55438eafec23a719d4ea2840f91d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 07:18:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/1346e19c3a2944ae96a0a49a24706913 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cf0eada1ac8ede04ee32c5eec0505088bcd38ba","revisions":{"9cf0eada1ac8ede04ee32c5eec0505088bcd38ba":{"kind":"REWORK","_number":1,"created":"2020-02-11 09:41:52.000000000","uploader":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"ref":"refs/changes/86/707086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/707086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/707086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/707086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/707086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/707086/1"}}},"commit":{"parents":[{"commit":"9ce6de3dfc384de6a78cf02f1aa917379f14109e","subject":"Merge \"Add a job to promote docs.airshipit.org content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ce6de3dfc384de6a78cf02f1aa917379f14109e"}]}],"author":{"name":"Neil Jerram","email":"neil@tigera.io","date":"2020-02-11 09:37:45.000000000","tz":0},"committer":{"name":"Neil Jerram","email":"neil@tigera.io","date":"2020-02-11 09:41:49.000000000","tz":0},"subject":"End gating for networking-calico, as retiring project from OpenStack","message":"End gating for networking-calico, as retiring project from OpenStack\n\nNote: project will continue elsewhere, but not under OpenStack\ngovernance.\n\nChange-Id: Ia13b651ce99ca295775187dc653e246b84fd1f43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cf0eada1ac8ede04ee32c5eec0505088bcd38ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cf0eada1ac8ede04ee32c5eec0505088bcd38ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
