)]}'
{"id":"openstack%2Fproject-config~707750","triplet_id":"openstack%2Fproject-config~master~I85438baf5bb31790a56fe5b38327361f0a2398e9","project":"openstack/project-config","branch":"master","topic":"venv-glean","hashtags":[],"change_id":"I85438baf5bb31790a56fe5b38327361f0a2398e9","subject":"nodepool-elements: Use venv for utilities","status":"MERGED","created":"2020-02-14 02:40:00.000000000","updated":"2020-03-10 23:34:49.000000000","submitted":"2020-03-10 23:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707750-1583883289142-77b288b9","meta_rev_id":"b9f08f73867e10326f00a8ff9c538509330cf032","_number":707750,"virtual_id_number":707750,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 23:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-03-10 23:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-10 23:21:44.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-03-10 23:21:44.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 23:14:58.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-10 23:21:44.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-03-10 23:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8acf31a288567a260712659e3c39c0f09b023a01","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-14 02:40:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"198fdddb4c5955462f6b33884ac4f1cb0efb8a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 02:49:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cab2999b3a0f4365b55a45cccec5b36d : SUCCESS in 7m 47s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/29f0f3b8ac29430cb9db222f0abd828b : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6637c5767de98e78c06f43802bfd3fa65ec7c929","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-14 02:58:17.000000000","message":"Patch Set 1: Workflow-1\n\nnote: this can not go in without further work to provide a backwards compatible pip/virtualenv environment.","accounts_in_message":[],"_revision_number":1},{"id":"2b7d6e4064995d9b84ff39fe037b750cf87447a5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-17 00:53:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"841967ca1fbfce80c1fd08f8bf918472532aa32b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 01:01:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3fae928420ca4c1988f270bcf2a9d7d3 : SUCCESS in 7m 45s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/5236e21e45fe47f88e664b5ff8de4106 : SUCCESS in 5m 08s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b2f8b6282cdc43c0a5669f4195f4a9e9 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8dfdc9db0788085c6a4f2a555b0e334204b7996a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-17 02:17:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4f7748e6b225c620d822765acd8bb3eb5e0be1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 02:28:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b017fa5d288c4e85919058afa40475a0 : SUCCESS in 10m 16s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/2fc07e152be345e1b3a62da9bf3cd005 : SUCCESS in 6m 40s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4e23d4dabff4445382b9ec9d66ab2620 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3},{"id":"1d7f1066466a5c3bffb9839ce1f19f011a4a57ba","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-18 19:27:21.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"83d5bf6d72f3e23767fe2c0fb9e8673feba223cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:50:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7d664d38a7f49649e81cf0742af835e : SUCCESS in 7m 56s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/69878798deee443589c69fc8f2512a2a : SUCCESS in 5m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e607cde03f394a7494d352cb347c8da0 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b798d429837fb9d5ab5a8d5fed184de61aa4798a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-10 22:51:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5ff8cf773709d7d7d2784c3231ed80399d515446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:03:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/760cd666671849eebc713183aff68ac6 : SUCCESS in 6m 41s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/bb696536b6ae47988f65434657b2b9ef : FAILURE in 3m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bf96e28aa73a439e85f747570295afc7 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":5},{"id":"b553a93581754ef305a996ffa3ca2ce0384ae720","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-10 23:12:18.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"cc544093bc6c46d532b6d4dcc961435fe8a9e999","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-10 23:14:58.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a4b3e7be0d52af2aaa74cb4a93d1ce616d6bb3a9","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-10 23:21:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c9198dced51e1f57076a83ef64943a270f31f082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:22:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de8c59d5ace74e4994e8d7004cbe5113 : SUCCESS in 6m 44s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6fceb2cc69f44e95bae1aa759682c009 : SUCCESS in 5m 10s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1308823dc7244e5d91b4aea32515defe : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":6},{"id":"967450617c804cf89cd56bdbcde496870f58cedb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:24:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"57261d07d1e6d3d1f93d8518ce4b9f8f7160050c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:34:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e904b66ffad34e98998169189d4b5d5e : SUCCESS in 6m 13s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/17fd4797e7884084aadf54bb866d462e : SUCCESS in 3m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0d2f4f701014451f9e8e18fce33657ee : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":6},{"id":"38085ebe2502d67af5bd58615a5bd5fd89605644","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 23:34:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7602343b6a7100059ed8f13fa00cb2ca7d34734c","revisions":{"704d1e82f30229676f514fdd483edda622e8b737":{"kind":"REWORK","_number":1,"created":"2020-02-14 02:40:00.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/1"}}},"commit":{"parents":[{"commit":"b393e477951ba3a38c63565c0824f4cf95ae292d","subject":"Merge \"check-release-approval: log Gerrit data on error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b393e477951ba3a38c63565c0824f4cf95ae292d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nNote this drops the dependency on pip-and-virtualenv element for\nnodepool elements.  Thus this should be integrated with changes to the\nbase jobs to ensure backwards-compatible pip and virtualenv\nenvironments that are currently expected by jobs is replicated before\nthey start.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704d1e82f30229676f514fdd483edda622e8b737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704d1e82f30229676f514fdd483edda622e8b737"}]},"branch":"refs/heads/master"},"f9a07c66626c43715d155c58c2deef062bc1cfbb":{"kind":"REWORK","_number":2,"created":"2020-02-17 00:53:22.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/2"}}},"commit":{"parents":[{"commit":"b393e477951ba3a38c63565c0824f4cf95ae292d","subject":"Merge \"check-release-approval: log Gerrit data on error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b393e477951ba3a38c63565c0824f4cf95ae292d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-17 00:39:50.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nSince we are no longer using pip to install anything during the\nbuilder, this drops the dependency on pip-and-virtualenv from\nnodepool-base.  Avoiding this element is our long-term goal, as it\u0027s\nmodification to system state are problematic in a number of ways.  To\nmaintain the status-quo, the pip-and-virtualenv element is added\nexplicitly to each build\u0027s element list, with a note on it\u0027s future.\n\nThe current plan for backwards compatability is to replicate the\nenvironment pip-and-virtualenv provides in a base role/job that can be\noptionally included.  To test this, provide a new node type\n\"ubuntu-bionic-plain\" that will not include the pip-and-virtualenv\nelement.  This is put on just one provider (rax) to minimise impact.\n\nThe dependent-change (and a dib release) is required for\nubuntu-bionic-plain to drop any dependency on pip-and-virtualenv, but\nthis change should be safe to merge at any point since it will not\nchange the status-quo for the existing images.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9a07c66626c43715d155c58c2deef062bc1cfbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9a07c66626c43715d155c58c2deef062bc1cfbb"}]},"branch":"refs/heads/master"},"d5e40841fd33eb562398ff70089ffa155febdfb3":{"kind":"REWORK","_number":3,"created":"2020-02-17 02:17:57.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/3"}}},"commit":{"parents":[{"commit":"b393e477951ba3a38c63565c0824f4cf95ae292d","subject":"Merge \"check-release-approval: log Gerrit data on error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b393e477951ba3a38c63565c0824f4cf95ae292d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-17 02:17:41.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nSince we are no longer using pip to install anything during the\nbuilder, this drops the dependency on pip-and-virtualenv from\nnodepool-base.  Avoiding this element is our long-term goal, as it\u0027s\nmodification to system state are problematic in a number of ways.  To\nmaintain the status-quo, the pip-and-virtualenv element is added\nexplicitly to each build\u0027s element list, with a note on it\u0027s future.\n\nThe current plan for backwards compatability is to replicate the\nenvironment pip-and-virtualenv provides in a base role/job that can be\noptionally included.  To test this, provide a new node type\n\"ubuntu-bionic-plain\" that will not include the pip-and-virtualenv\nelement.  This is put on just one provider (rax) to minimise impact.\n\nThe dependent-change (and a dib release) is required for\nubuntu-bionic-plain to drop any dependency on pip-and-virtualenv, but\nthis change should be safe to merge at any point since it will not\nchange the status-quo for the existing images.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d5e40841fd33eb562398ff70089ffa155febdfb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d5e40841fd33eb562398ff70089ffa155febdfb3"}]},"branch":"refs/heads/master"},"405c2c2572d77420360a4ba6f07e9d5da817b41b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-18 19:27:21.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/4"}}},"commit":{"parents":[{"commit":"b393e477951ba3a38c63565c0824f4cf95ae292d","subject":"Merge \"check-release-approval: log Gerrit data on error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b393e477951ba3a38c63565c0824f4cf95ae292d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-18 19:26:25.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nSince we are no longer using pip to install anything during the\nbuilder, this drops the dependency on pip-and-virtualenv from\nnodepool-base.  Avoiding this element is our long-term goal, as it\u0027s\nmodification to system state are problematic in a number of ways.  To\nmaintain the status-quo, the pip-and-virtualenv element is added\nexplicitly to each build\u0027s element list, with a note on it\u0027s future.\n\nThe current plan for backwards compatability is to replicate the\nenvironment pip-and-virtualenv provides in a base role/job that can be\noptionally included.  To test this, provide a new node type\n\"ubuntu-bionic-plain\" that will not include the pip-and-virtualenv\nelement.  This is put on just one provider (rax) to minimise impact.\n\nThe dependent-change (and a dib release) is required before merge so\nthe ensure-venv element is available.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/405c2c2572d77420360a4ba6f07e9d5da817b41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/405c2c2572d77420360a4ba6f07e9d5da817b41b"}]},"branch":"refs/heads/master"},"0a13c5bc3893303af65c5700aed8ce647d546bdb":{"kind":"REWORK","_number":5,"created":"2020-03-10 22:51:41.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/5"}}},"commit":{"parents":[{"commit":"cb6e5eef3892e0299730ede1749dff51ed191cde","subject":"Merge \"Remove Fedora 29 from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb6e5eef3892e0299730ede1749dff51ed191cde"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-10 22:51:21.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nSince we are no longer using pip to install anything during the\nbuilder, this drops the dependency on pip-and-virtualenv from\nnodepool-base.  Avoiding this element is our long-term goal, as it\u0027s\nmodification to system state are problematic in a number of ways.  To\nmaintain the status-quo, the pip-and-virtualenv element is added\nexplicitly to each build\u0027s element list, with a note on it\u0027s future.\n\nThe current plan for backwards compatability is to replicate the\nenvironment pip-and-virtualenv provides in a base role/job that can be\noptionally included.  To test this, provide a new node type\n\"ubuntu-bionic-plain\" that will not include the pip-and-virtualenv\nelement.  This is put on just one provider (rax) to minimise impact.\n\nThe dependent-change (and a dib release) is required before merge so\nthe ensure-venv element is available.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a13c5bc3893303af65c5700aed8ce647d546bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a13c5bc3893303af65c5700aed8ce647d546bdb"}]},"branch":"refs/heads/master"},"7602343b6a7100059ed8f13fa00cb2ca7d34734c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-03-10 23:12:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/50/707750/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/707750/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/707750/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/707750/6"}}},"commit":{"parents":[{"commit":"7801ad35a680ce83d9586af2edbe6d0726269a5a","subject":"Fix new dib-lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7801ad35a680ce83d9586af2edbe6d0726269a5a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 02:26:10.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-10 23:12:00.000000000","tz":660},"subject":"nodepool-elements: Use venv for utilities","message":"nodepool-elements: Use venv for utilities\n\nSince all platforms have Python 3, use the new ensure-venv element\nfrom the dependent change to install bindep and os-testr.\n\nSince we are no longer using pip to install anything during the\nbuilder, this drops the dependency on pip-and-virtualenv from\nnodepool-base.  Avoiding this element is our long-term goal, as it\u0027s\nmodification to system state are problematic in a number of ways.  To\nmaintain the status-quo, the pip-and-virtualenv element is added\nexplicitly to each build\u0027s element list, with a note on it\u0027s future.\n\nThe current plan for backwards compatability is to replicate the\nenvironment pip-and-virtualenv provides in a base role/job that can be\noptionally included.  To test this, provide a new node type\n\"ubuntu-bionic-plain\" that will not include the pip-and-virtualenv\nelement.  This is put on just one provider (rax) to minimise impact.\n\nThe dependent-change (and a dib release) is required before merge so\nthe ensure-venv element is available.\n\nDepends-On: https://review.opendev.org/707513\nChange-Id: I85438baf5bb31790a56fe5b38327361f0a2398e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7602343b6a7100059ed8f13fa00cb2ca7d34734c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7602343b6a7100059ed8f13fa00cb2ca7d34734c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
