)]}'
{"id":"openstack%2Fproject-config~708023","triplet_id":"openstack%2Fproject-config~master~I7ec0196ef9914b821cb007bc184252393554d243","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I7ec0196ef9914b821cb007bc184252393554d243","subject":"Add polymer-bridges project from gerrit","status":"MERGED","created":"2020-02-16 15:37:22.000000000","updated":"2020-02-17 14:57:30.000000000","submitted":"2020-02-17 14:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708023-1581950851760-824f8b14","meta_rev_id":"40d67aeb727a62d69bbc48a8f1d76db4573e31f9","_number":708023,"virtual_id_number":708023,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 14:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-02-16 15:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-17 14:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-02-17 14:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 15:38:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-17 14:36:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-02-17 14:47:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e02d02e1ab73f867e4d6c8de082fe1eef304e80a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-16 15:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fc3421c2b5d422ea326651281f25bf89b8bb90c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-16 15:38:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d14244b3353c2a7e6d60620d77c1bb8884f2a5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 15:45:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6e1cafc089ca4d2cada0ff37bc79bf98 : SUCCESS in 8m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/43e90275097e4cb48155d5b0c6b5cdd2 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a1b7240203c434f5820aae6cadd42ea6d7fe2021","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-17 14:36:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e33ad626c8cdaf42911485d99d0951cfcffec5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ce35e03b9d137fcebe2d5f6c106b63c4cb270f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:47:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/559f9df048e74e6e96112b1523051507 : SUCCESS in 8m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0b3c7fbd3b2e4b87bbfb7dbe590ea445 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6df075936e11992c15a55b44f1a9b0d09e34f68c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:47:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"40d67aeb727a62d69bbc48a8f1d76db4573e31f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:57:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- project-config-test-artifacts-to-afs https://zuul.opendev.org/t/openstack/build/fc34c2ecd44a442d92e7047b917393e3 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7c6f30ed96fdbc5a41255682d62af541e0dee05","revisions":{"f7c6f30ed96fdbc5a41255682d62af541e0dee05":{"kind":"REWORK","_number":1,"created":"2020-02-16 15:37:22.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/23/708023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/708023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/708023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/708023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/708023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/708023/1"}}},"commit":{"parents":[{"commit":"b393e477951ba3a38c63565c0824f4cf95ae292d","subject":"Merge \"check-release-approval: log Gerrit data on error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b393e477951ba3a38c63565c0824f4cf95ae292d"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-16 15:36:45.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-02-16 15:36:45.000000000","tz":-360},"subject":"Add polymer-bridges project from gerrit","message":"Add polymer-bridges project from gerrit\n\nIt\u0027s a new submodule needed by gerrit, so we need it too.\n\nChange-Id: I7ec0196ef9914b821cb007bc184252393554d243\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7c6f30ed96fdbc5a41255682d62af541e0dee05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7c6f30ed96fdbc5a41255682d62af541e0dee05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
