)]}'
{"id":"openstack%2Fproject-config~708961","triplet_id":"openstack%2Fproject-config~master~I20db53eeeea07e9d0a8cf5ea9c24f790f39db9f0","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I20db53eeeea07e9d0a8cf5ea9c24f790f39db9f0","subject":"Create openinfralabs/contrib repo","status":"MERGED","created":"2020-02-20 20:20:29.000000000","updated":"2020-02-21 22:29:38.000000000","submitted":"2020-02-21 22:29:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708961-1582324178049-5f1f21a0","meta_rev_id":"a14b862eeff372a24e04b3e95776608d7cfa594f","_number":708961,"virtual_id_number":708961,"owner":{"_account_id":31651,"name":"Michael Daitzman","email":"msd@bu.edu","username":"msdisme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 22:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-02-21 19:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-20 21:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-21 22:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 20:20:40.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-02-20 20:53:40.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-02-21 19:04:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-21 22:21:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-02-21 22:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f02c676545369c423e1f4eba236c38a115d2db19","author":{"_account_id":31651,"name":"Michael Daitzman","email":"msd@bu.edu","username":"msdisme"},"date":"2020-02-20 20:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"004c8e298cc53f78150a0730fe798abae181119f","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-02-20 20:20:40.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"0b0c815bb07dbce9254fe0dfeeb8a35979d53e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 20:30:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f33f8d3e0ec14af1b59ef8083c37774f : SUCCESS in 7m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/42294a646ca7419d8521d786a97d35a0 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b6270e5439afe9f347953bfc6977c05f1aa846b9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-20 20:37:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e93e3fd59a0e2e731d441fe6a4f1cd44acaedd85","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-20 20:45:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0eccc54aa8e8cfd030b9e3576ea169385f5ec64","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-20 20:47:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0363ff21646ffa04d19cc4d81f4094bc0744af35","author":{"_account_id":31651,"name":"Michael Daitzman","email":"msd@bu.edu","username":"msdisme"},"date":"2020-02-20 21:22:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3f1d4cbb1fc70a75a59d692b060fb32cc82cc18","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-20 21:24:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"970ef3ac098b7801d165b663728afcddaab8c84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 21:31:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ce1ae1f45b5646baa2cdb74bec11ca2c : SUCCESS in 7m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aa0aadfb7efd46b6aa8991b4398734a9 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1f245ff965910e0f440ba6312b841d5fcb0a37d5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-21 19:03:13.000000000","message":"Patch Set 2:\n\nMichael, do you have an initial repo to import with history - or are you fine to start with an empty repo?","accounts_in_message":[],"_revision_number":2},{"id":"1de3da743780893cb4b91c4e3a47d36eb5dfe9b7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-21 19:04:42.000000000","message":"Patch Set 2: Code-Review+2\n\nchange is fine - will approve once I have an answer to my question.","accounts_in_message":[],"_revision_number":2},{"id":"d51f518befe676a0714450e79704c418ee7a09aa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-21 22:21:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f6c1c14d5975a6feffa95ae7fef42ec35430ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 22:21:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e62af1d038ad876f89024a6a1fafb0f7ad995326","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-21 22:22:26.000000000","message":"Patch Set 2:\n\nMichael replied via E-mail: \"Empty is great for now!\"","accounts_in_message":[],"_revision_number":2},{"id":"febbf8dec60df140d0019114ea3dd896edd7c779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 22:29:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b9456ff131354c84bf41a42d68659b0f : SUCCESS in 7m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b238ed38ceb8489e8766d5c9b43a02bf : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"01ba0f3ff884319bbb8f5fae36642cb51c0412d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 22:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e82cb0955ed76c939a5b57075e8612aba04998ae","revisions":{"f1b5d19d077961811c877f16bf94b91e213648ee":{"kind":"REWORK","_number":1,"created":"2020-02-20 20:20:29.000000000","uploader":{"_account_id":31651,"name":"Michael Daitzman","email":"msd@bu.edu","username":"msdisme"},"ref":"refs/changes/61/708961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/708961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/708961/1"}}},"commit":{"parents":[{"commit":"3e0549776b6f243566ce643fd226a1a8b61325d1","subject":"Merge \"Remove tarball artifact publishing to static site\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e0549776b6f243566ce643fd226a1a8b61325d1"}]}],"author":{"name":"msdisme","email":"mdaitzmane@gmail.com","date":"2020-02-20 20:16:01.000000000","tz":-300},"committer":{"name":"msdisme","email":"msd@bu.edu","date":"2020-02-20 20:19:42.000000000","tz":-300},"subject":"Create openinfralabs/contrib repo","message":"Create openinfralabs/contrib repo\n\nThis is a place for OpenInfraLabs.org contributors to share code, scripts, etc.\n\nChange-Id: I20db53eeeea07e9d0a8cf5ea9c24f790f39db9f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f1b5d19d077961811c877f16bf94b91e213648ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f1b5d19d077961811c877f16bf94b91e213648ee"}]},"branch":"refs/heads/master"},"e82cb0955ed76c939a5b57075e8612aba04998ae":{"kind":"REWORK","_number":2,"created":"2020-02-20 21:22:21.000000000","uploader":{"_account_id":31651,"name":"Michael Daitzman","email":"msd@bu.edu","username":"msdisme"},"ref":"refs/changes/61/708961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/708961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/708961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/708961/2"}}},"commit":{"parents":[{"commit":"3e0549776b6f243566ce643fd226a1a8b61325d1","subject":"Merge \"Remove tarball artifact publishing to static site\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e0549776b6f243566ce643fd226a1a8b61325d1"}]}],"author":{"name":"msdisme","email":"mdaitzmane@gmail.com","date":"2020-02-20 20:16:01.000000000","tz":-300},"committer":{"name":"msdisme","email":"msd@bu.edu","date":"2020-02-20 21:21:56.000000000","tz":-300},"subject":"Create openinfralabs/contrib repo","message":"Create openinfralabs/contrib repo\n\nThis is a place for OpenInfraLabs.org contributors to share code, scripts, etc.\n\nChange-Id: I20db53eeeea07e9d0a8cf5ea9c24f790f39db9f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e82cb0955ed76c939a5b57075e8612aba04998ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e82cb0955ed76c939a5b57075e8612aba04998ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
