)]}'
{"id":"openstack%2Fproject-config~710110","triplet_id":"openstack%2Fproject-config~master~Id6e2ddcc3cbced430f73efdf03b6425cf043209a","project":"openstack/project-config","branch":"master","topic":"opendev-puppet-forge","hashtags":[],"change_id":"Id6e2ddcc3cbced430f73efdf03b6425cf043209a","subject":"Remove puppet-forge upload jobs from opendev","status":"MERGED","created":"2020-02-26 18:40:27.000000000","updated":"2020-02-26 19:43:51.000000000","submitted":"2020-02-26 19:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710110-1582746231924-3bf77f9f","meta_rev_id":"012dc036ed47259a6eb9742e55c31c13ca2247fa","_number":710110,"virtual_id_number":710110,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 19:43:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 19:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-26 19:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 19:28:04.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-02-26 19:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bb12a4b65a941bf4b9b812ae848e1cc1afb2302","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-26 18:40:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7376523166befa5749b87ff2ba4f5a3590835a1b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-26 18:42:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAnd as far as I know, we never did.","accounts_in_message":[],"_revision_number":1},{"id":"51e7c1456dc961f8e977a8326dc6f269e0c9fd94","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-26 18:45:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f4efb93419b62006d69b4cec0ac6ad3960646fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 18:54:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5173adbf18c45a59f2c7b1efb533d92 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2},{"id":"df468ac28f259a6d43c769ebb9769fa8b82a4ec9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-26 19:28:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"80e76dbbcbbc8860c826f323e3ec48f37bd61719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:28:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d58624d1c42591d26ec5c90b94c19708e7d0a4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:43:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"523fef3bb257ecf2b9146afabcc3acaf1a49aeb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:43:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fde2e0ea783f49f7a5c10f72468f5c23 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"962cb9ad3c5fbdc3c4a4b306eebf68fada5d17ae","revisions":{"a6beb6232a1f2bb2ea3d26a9e80c7fdf3fabc311":{"kind":"REWORK","_number":1,"created":"2020-02-26 18:40:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/710110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/710110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/710110/1"}}},"commit":{"parents":[{"commit":"37b9019e963e3a77e1a1cf9e8ab1d25ccaba530a","subject":"Make update_constraints.sh aware of python_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37b9019e963e3a77e1a1cf9e8ab1d25ccaba530a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:39:59.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:39:59.000000000","tz":60},"subject":"Remove legacy-puppet-openstack-integration job","message":"Remove legacy-puppet-openstack-integration job\n\nRemove this job from puppet-storyboard, we do not upload anymore to the\nforge.\n\nChange-Id: Id6e2ddcc3cbced430f73efdf03b6425cf043209a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6beb6232a1f2bb2ea3d26a9e80c7fdf3fabc311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6beb6232a1f2bb2ea3d26a9e80c7fdf3fabc311"}]},"branch":"refs/heads/master"},"962cb9ad3c5fbdc3c4a4b306eebf68fada5d17ae":{"kind":"REWORK","_number":2,"created":"2020-02-26 18:45:00.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/710110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/710110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/710110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/710110/2"}}},"commit":{"parents":[{"commit":"37b9019e963e3a77e1a1cf9e8ab1d25ccaba530a","subject":"Make update_constraints.sh aware of python_versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37b9019e963e3a77e1a1cf9e8ab1d25ccaba530a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:39:59.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 18:44:41.000000000","tz":60},"subject":"Remove puppet-forge upload jobs from opendev","message":"Remove puppet-forge upload jobs from opendev\n\nWe do not uplload from opendev to puppet forge anymore:\n\nRemove usage of puppet-forge uploads jobs from opendev. This removes\nthe useage of release jobs:\n- legacy-puppet-storyboard-forge-upload from puppet-storyboard\n- legacy-puppet-httpd-forge-upload from puppet-httpd\n\nChange-Id: Id6e2ddcc3cbced430f73efdf03b6425cf043209a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/962cb9ad3c5fbdc3c4a4b306eebf68fada5d17ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/962cb9ad3c5fbdc3c4a4b306eebf68fada5d17ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
