)]}'
{"id":"openstack%2Fproject-config~710700","triplet_id":"openstack%2Fproject-config~master~Ie500bd0d58d0616240b627d1328be36087467864","project":"openstack/project-config","branch":"master","topic":"nodepool-scripts","hashtags":[],"change_id":"Ie500bd0d58d0616240b627d1328be36087467864","subject":"nodepool: Add nb01-test.opendev.org config file","status":"MERGED","created":"2020-03-02 06:10:47.000000000","updated":"2020-03-03 02:13:15.000000000","submitted":"2020-03-03 02:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710700-1583201595350-d353a4dd","meta_rev_id":"ac4f7a244bdc40ec4409a9e2bd3ece9e8cb8a677","_number":710700,"virtual_id_number":710700,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 02:13:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-03-03 01:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-03 02:03:54.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-03 02:03:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 07:11:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-02 17:20:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-03 01:56:23.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-03 02:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c18a6a48b72e7a7674403d4b41317f52ea5dac31","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-02 06:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96e2caaa39447975bb57b285287db5103770a987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 06:22:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca816dfc6afa415cbfde0d180bd160b8 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3af6bd5848a95f4cd6a4999f6948f3a09d3846e3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-02 07:11:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96d976b73efe065091c21d9ce3b0762e65030c56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 17:20:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"970c752b64f79d7bd5c9ad1045b31b0136e42116","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-03 00:06:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6a0f81ebbecf0ea9b6970461294469157cc730a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:21:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3cd1009c1aae4f2b96c43d15903a240c : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dfcea2bc24a36f8391d575a96e7a9ce42dfa3773","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-03 01:56:23.000000000","message":"Patch Set 2: Code-Review+2\n\nNote this repo is a trusted repo in our zuul tenant config. That means we won\u0027t be able to test changes to this file as they are proposed. We\u0027ll be able to test it just fine after merging though.","accounts_in_message":[],"_revision_number":2},{"id":"47fb93165e61146255a6faf19a15d03c17a1b59a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-03 02:03:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\ntaking ajaeger\u0027s prior vote as i only updated the needed-by","accounts_in_message":[],"_revision_number":2},{"id":"2a3a038d6e3cd749fa9367d5de361d61a97ac4a8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-03 02:04:02.000000000","message":"Removed Code-Review+2 by Ian Wienand \u003ciwienand@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"7c72b741646bdc453f5700b87da7b3e70689dacd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:04:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee134d7b161fe6f125a6646657f851001277e44a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:13:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/692777c40dc6472b901f59499b5fe9a1 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"161e9d797b6936f1fa4dec93bc547253a7d3b27e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:13:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79c18df68b400efeb6bf8749c30697e49b962518","revisions":{"97a9fbe63e8ee05bb6ef9c64adbefbda9e53ed3b":{"kind":"REWORK","_number":1,"created":"2020-03-02 06:10:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/00/710700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/710700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/710700/1"}}},"commit":{"parents":[{"commit":"e674d5b4b1dd5fbf9b532d5b6fd5669d15208659","subject":"Remove nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e674d5b4b1dd5fbf9b532d5b6fd5669d15208659"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-02 05:35:12.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-02 05:35:12.000000000","tz":660},"subject":"nodepool: Add nb01-test.opendev.org config file","message":"nodepool: Add nb01-test.opendev.org config file\n\nThis config file is only used by the system-config-run-nodepool test\nin the gate.  We want to keep this config file in project-config so\nthat we can deploy all the configuration parts in that test without\nspecial paths and as if nb01-test.opendev.org was a production host.\n\nChange-Id: Ie500bd0d58d0616240b627d1328be36087467864\nNeeded-By: TBC\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/97a9fbe63e8ee05bb6ef9c64adbefbda9e53ed3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/97a9fbe63e8ee05bb6ef9c64adbefbda9e53ed3b"}]},"branch":"refs/heads/master"},"79c18df68b400efeb6bf8749c30697e49b962518":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-03 00:06:49.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/00/710700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/710700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/710700/2"}}},"commit":{"parents":[{"commit":"e674d5b4b1dd5fbf9b532d5b6fd5669d15208659","subject":"Remove nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e674d5b4b1dd5fbf9b532d5b6fd5669d15208659"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-02 05:35:12.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-03 00:06:05.000000000","tz":660},"subject":"nodepool: Add nb01-test.opendev.org config file","message":"nodepool: Add nb01-test.opendev.org config file\n\nThis config file is only used by the system-config-run-nodepool test\nin the gate.  We want to keep this config file in project-config so\nthat we can deploy all the configuration parts in that test without\nspecial paths and as if nb01-test.opendev.org was a production host.\n\nChange-Id: Ie500bd0d58d0616240b627d1328be36087467864\nNeeded-By: https://review.opendev.org/710888\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/79c18df68b400efeb6bf8749c30697e49b962518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/79c18df68b400efeb6bf8749c30697e49b962518"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
