)]}'
{"id":"openstack%2Fproject-config~710706","triplet_id":"openstack%2Fproject-config~master~If063c60820655a980305f5c47915eaf91f9dbfca","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"If063c60820655a980305f5c47915eaf91f9dbfca","subject":"Reduce Linaro London node to 0 as test needed","status":"MERGED","created":"2020-03-02 06:36:03.000000000","updated":"2020-03-02 07:03:25.000000000","submitted":"2020-03-02 07:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710706-1583132605463-1d1e633c","meta_rev_id":"54271ee812384380baca4940067461420a695e5b","_number":710706,"virtual_id_number":710706,"owner":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 07:03:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-02 06:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-02 06:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 06:37:25.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-03-02 07:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f812d21b4d70d6058ef25a618c4524b15b27ffc3","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-03-02 06:36:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2101d057a58a7dfde060eb57ff0a3863db5af75","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-02 06:37:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfast approving for testing purposes","accounts_in_message":[],"_revision_number":1},{"id":"1bc518077ac0cd52d327e7631a6292714edeba2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 06:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/38134ed4dd174e5e9929d9ed9778e4d2 : SUCCESS in 12m 14s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ea7dd67bc8304e9a9f202bd77948e31a : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"34668ad284d9842e6c8465b86a0e65cc25130eba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 06:53:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a27ca46bb327a45599c1f7664859058656b2686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 07:03:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54271ee812384380baca4940067461420a695e5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 07:03:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f84baa23e088462f8c8ee4dc67f9b891 : SUCCESS in 9m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4f6a9c6abf6b421eaf27569bb1ccb83c : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c37c64d63805901e1b208f7f6e1dd8901355fb96","revisions":{"c37c64d63805901e1b208f7f6e1dd8901355fb96":{"kind":"REWORK","_number":1,"created":"2020-03-02 06:36:03.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/06/710706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/710706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/710706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/710706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/710706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/710706/1"}}},"commit":{"parents":[{"commit":"e67cadc91881428cd92728bdb6a0f752dedfca2d","subject":"reduce machine number from Linaro London","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e67cadc91881428cd92728bdb6a0f752dedfca2d"}]}],"author":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-03-02 06:35:48.000000000","tz":480},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-03-02 06:35:48.000000000","tz":480},"subject":"Reduce Linaro London node to 0 as test needed","message":"Reduce Linaro London node to 0 as test needed\n\nChange-Id: If063c60820655a980305f5c47915eaf91f9dbfca\nSigned-off-by: Kevin Zhao \u003ckevin.zhao@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c37c64d63805901e1b208f7f6e1dd8901355fb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c37c64d63805901e1b208f7f6e1dd8901355fb96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
