)]}'
{"id":"openstack%2Fproject-config~710900","triplet_id":"openstack%2Fproject-config~master~I8b82950237ef69c4941866900cac9bda42f58ca2","project":"openstack/project-config","branch":"master","topic":"nodepool-scripts","hashtags":[],"change_id":"I8b82950237ef69c4941866900cac9bda42f58ca2","subject":"Restore blank nodepool/scripts","status":"MERGED","created":"2020-03-03 04:01:39.000000000","updated":"2020-03-03 06:32:18.000000000","submitted":"2020-03-03 06:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710900-1583217138367-bd11f162","meta_rev_id":"edae13ad4806bd98e9439e37c905e5e99099fcbd","_number":710900,"virtual_id_number":710900,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 06:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-03 04:05:54.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-03-03 06:12:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-03-03 06:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 06:17:25.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-03 06:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f2c3a53359151734d2189e75412771dfdeb10d0","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-03 04:01:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f71274fef80e1ed979996c360cdc6e372ccb59d2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-03 04:05:54.000000000","message":"Patch Set 1:\n\ni believe this is the issue with the system-config gate\n\nhttps://zuul.opendev.org/t/openstack/build/3375d15cc9994eb2b9ba20e4d9fd0483/log/job-output.txt#6126\n\nError: /Stage[main]/Nodepool/File[/etc/nodepool/scripts]: Could not evaluate: Could not retrieve information from environment production source(s) file:/etc/project-config/nodepool/scripts","accounts_in_message":[],"_revision_number":1},{"id":"933a69cd6b74300fb704ddb6628de9e55142afe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 04:11:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6582feaad2d745e68196486d107fe5f6 : SUCCESS in 8m 32s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/8536d4657cc14c5f926377bf06c6681b : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7cd558118a9548d8fb88004a85a38981ac645be5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-03 06:12:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f0dbbc31fee6e61c5a34cf83e29e1a7ea246707","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-03 06:17:25.000000000","message":"Patch Set 1: Workflow+1\n\nsingle core approving to unbreak system-config and friends","accounts_in_message":[],"_revision_number":1},{"id":"1e23f1e715171bb122b0525e36994ffa7c1188f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:17:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0dc017a7eba1b1dec1612d23b935924f519e9e26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:32:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/354729e2d3f742d69b59aa6fdd1713de : SUCCESS in 12m 18s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/2fb4a27d25d24791b6f6a84efadbd968 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"313bfdeedb42711721b9c359d7298bf31b23e5a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:32:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7","revisions":{"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7":{"kind":"REWORK","_number":1,"created":"2020-03-03 04:01:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/00/710900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/710900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/710900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/710900/1"}}},"commit":{"parents":[{"commit":"48b291f674a287fd5aefeb8f70f22c2c472ad395","subject":"Merge \"nodepool: Add nb01-test.opendev.org config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/48b291f674a287fd5aefeb8f70f22c2c472ad395"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-03 03:57:12.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-03 04:01:23.000000000","tz":660},"subject":"Restore blank nodepool/scripts","message":"Restore blank nodepool/scripts\n\nI was a little too hasty in I76acbd08acda20c88ff9fd9148e3918b78d6c6c6\nremoving the scripts/ directory.  It has broken the puppeting of old\nhosts.\n\nRestore the directory with a blank file explaining the situation.\n\nAlso, we don\u0027t need to copy this in the nodepool elements; remove that.\n\nChange-Id: I8b82950237ef69c4941866900cac9bda42f58ca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
