)]}'
{"id":"openstack%2Fproject-config~711114","triplet_id":"openstack%2Fproject-config~master~I8e0ee3829a9d603c09f146f2a4a9ddc39656b885","project":"openstack/project-config","branch":"master","topic":"tripleo-ipa","hashtags":[],"change_id":"I8e0ee3829a9d603c09f146f2a4a9ddc39656b885","subject":"Add a new project and repository for tripleo-ipa","status":"MERGED","created":"2020-03-03 21:58:18.000000000","updated":"2020-03-13 15:44:45.000000000","submitted":"2020-03-13 15:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711114-1584114285347-c6fdaf72","meta_rev_id":"c54b3a7b9147fa84b121bfe288959ef58493dd05","_number":711114,"virtual_id_number":711114,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"date":"2020-03-11 00:52:53.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 15:44:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-13 15:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2020-03-12 07:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-10 14:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2020-03-10 15:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-03-10 18:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-11 15:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-13 15:34:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 21:58:18.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-03-05 23:45:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 14:21:34.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2020-03-10 14:28:33.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2020-03-10 15:44:12.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-03-10 18:30:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-11 15:51:46.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-03-12 07:26:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-13 15:34:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-13 15:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3fcf30242f71e394a46f9c117156afcdeb3e9e1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-03 21:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4db07669fa5facf535df0a6b41784a72d8b71a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6fb138d9f04b4672a36cb6dd2769ce0f : SUCCESS in 8m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f6d215ddd78741148e80097ae0893d37 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b136d3c0737d36f92945e1bf22da163fcc1e4c45","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-04 05:18:01.000000000","message":"Patch Set 1:\n\nthis misses a change for governance repository, please push it as well.","accounts_in_message":[],"_revision_number":1},{"id":"950f346b6fee5c4c67005510db786e0b7624538d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-04 07:19:32.000000000","message":"Patch Set 1:\n\nNote that you will start with an empty repository. Your very first\npatch needs to setup the repository and add all infrastructure so that\nthe jobs pass (this can be done with a cookiecutter project).\n\nAlso, you can import an existing repository including it\u0027s history if\nyou have one. Once the repository is created, importing is not\npossible anymore, changes will have to go through the normal review\nprocess.\n\nIf you have a lot of history to import, please use the upstream field\ninstead of creating a repository and then pushing the patches one at a\ntime. Pushing a large number of related patches all at one time causes\nthe CI infrastructure to slow down, which impacts work on all of the\nother projects using it. After repo creation the infra team will not\npush any content to it; this is your only chance to import directly.\n\nPlease confirm that starting with an empty repo is fine.","accounts_in_message":[],"_revision_number":1},{"id":"6c2458bd9147c7b22d9baa6d4712b0004437d5cb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-04 15:11:58.000000000","message":"Patch Set 1:\n\n\u003e Note that you will start with an empty repository. Your very first\n \u003e patch needs to setup the repository and add all infrastructure so\n \u003e that\n \u003e the jobs pass (this can be done with a cookiecutter project).\n \u003e \n \u003e Also, you can import an existing repository including it\u0027s history\n \u003e if\n \u003e you have one. Once the repository is created, importing is not\n \u003e possible anymore, changes will have to go through the normal review\n \u003e process.\n \u003e \n \u003e If you have a lot of history to import, please use the upstream\n \u003e field\n \u003e instead of creating a repository and then pushing the patches one\n \u003e at a\n \u003e time. Pushing a large number of related patches all at one time\n \u003e causes\n \u003e the CI infrastructure to slow down, which impacts work on all of\n \u003e the\n \u003e other projects using it. After repo creation the infra team will\n \u003e not\n \u003e push any content to it; this is your only chance to import\n \u003e directly.\n \u003e \n \u003e Please confirm that starting with an empty repo is fine.\n\nThanks for the information, Andreas.\n\nI\u0027ll provide some additional context. This ansible role is effectively replacing x/novajoin [0]. We\u0027d like to start with a fresh repository. We don\u0027t have an existing project we\u0027re going to import.\n\nI don\u0027t see novajoin as a project in governance. As far as I know, we\u0027re not targeting this to be OpenStack official. Do I still need to propose a patch to openstack/governance?\n\n[0] https://opendev.org/x/novajoin","accounts_in_message":[],"_revision_number":1},{"id":"6b15a1ae71680e18a9d025d76ce81da1b1dcd3af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-04 15:27:39.000000000","message":"Patch Set 1: Code-Review-1\n\nIf this is not under governance, submit another namespace, it cannot go into openstack anymore.","accounts_in_message":[],"_revision_number":1},{"id":"a082e3bc9548752db9bdabda6f5b4af0d08a84df","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-04 15:39:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf797faa793771b888e01401cdb2be34419f6e0d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-04 15:40:30.000000000","message":"Patch Set 2:\n\n\u003e If this is not under governance, submit another namespace, it\n \u003e cannot go into openstack anymore.\n\nDone. Thanks for the clarification, Andreas.","accounts_in_message":[],"_revision_number":2},{"id":"51a4bad784cfe6898e8397a230b26bd90afc5690","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-04 15:42:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2820f728d772f3c85b539bf99c35bb1715477129","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-04 15:43:10.000000000","message":"Patch Set 2:\n\nthis was discussed and approved in the last tripleo community meeting.","accounts_in_message":[],"_revision_number":2},{"id":"61d587602946f1a3a9bab2f5b47b8791e516b4ec","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-04 15:47:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdb30b78446a057bc2f50592e171be7c9db5f0e4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-04 15:48:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"021ba5a1da81aa430d8a0da06b91e07528338705","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-04 15:49:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3301485c6a2a656318d32543dcb6b3e925154aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:50:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d1c23f6a3f3f49b59f4c544fa92d8b5c : SUCCESS in 8m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/311f4aee807144f980b8483c029d01f7 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4a919f3b6fd9c45abc432648992639e5bb3c74e6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-04 15:51:09.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acfcd57b2e1dba60bc57ee1ca234e0794ad86d50","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-03-04 16:24:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6c9a78b7889dbe099863e8a04b18f0198cf5ab6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-04 17:48:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46758d6d219348363d6fbac5baa9b526b9969342","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-03-04 20:14:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6cedd30d6f9f9649efa6a70d55ef7eafd6f2ffe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-04 20:32:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b890bbfc51cf143c35afb6ec35586c31e4921f34","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-03-04 21:26:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22ebcb3b81f433f4e0f0e7fd0f5fef1fcb55deba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-04 21:48:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"239486a6cb328d93dc1bf976de57e5bfaf542cc9","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-03-04 22:08:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"42c65b77d9487bf107e406013208e0063b26f8ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-05 00:42:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f6ffd1c329ed87c22e68befbf95639b2616ec8a","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-03-05 02:49:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed0fea8016e2b85ccc3f016c567a71a941aa483f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-05 23:45:33.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cccf92a5c91e18d41a21336a028a1b35e0153763","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-05 23:52:20.000000000","message":"Patch Set 2: -Code-Review\n\nK i\u0027ll drop my objections as long as we are looking to properly integrate the upstream ipa role(s)","accounts_in_message":[],"_revision_number":2},{"id":"346066b2f822a7040af273166fdf45a2a9e11034","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-06 14:43:20.000000000","message":"Patch Set 2:\n\n\u003e K i\u0027ll drop my objections as long as we are looking to properly\n \u003e integrate the upstream ipa role(s)\n\nOk - I\u0027d like to clarify which option we agree on:\n\n1.) Are we agreeing that we will create x/tripleo-ipa, invoke it from tripleo-ansible or tripleo-heat-templates for tls-enabled deployments? It will be separately built, maintained, and packaged from tripleo-ansible and tripleo-heat-templates.\n\n2.) Or are we saying that we no longer need to create x/tripleo-ipa since we\u0027ve agreed that the primary responsibilities would make sense in tripleo-ansible assuming we upstream the necessary gaps in freeipa/ansible-freeipa and create appropriate filters?","accounts_in_message":[],"_revision_number":2},{"id":"12daebd657ba602efd8319dc55f15019f5868fd5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-10 13:48:46.000000000","message":"Patch Set 2:\n\nwe\u0027re going with #1","accounts_in_message":[],"_revision_number":2},{"id":"62145990372d652a8db54167455837b193a7a9f1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2020-03-10 13:52:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"01d318109c95c19906c8c69a47491f09056e93d6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-10 13:54:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67b7b9e72d3478bfcfc0b140b7d13e24a6a2d76f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-10 13:54:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b5f9cf3ee39cd515055dd4dcbb37dc82e8435b5","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2020-03-10 13:54:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c59e07256533b583f90b86be6df96ec6a74f6c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:08:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f55cc591f0d84ee994b61d3ea5da544e : SUCCESS in 8m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2dbf10e6d28a439c8991accd95e32b9e : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"a4249482087a8d292f2eccd6b81ef9101be18f39","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-10 14:20:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"68fc55b09ac649a92d67b298e79994ac0ae8a681","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-03-10 14:21:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c4899391d4311bdeb60e1c7d6561ee3046099bb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-10 14:25:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6a816efbe313b55a20c4d6e57de04ab87b8e69d9","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2020-03-10 14:28:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7f5d93e157e9b2c59bcea35f829534146487ff5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:43:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e1d64f737cf24c8f92ace47bfc1c476b : SUCCESS in 7m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cecbc0b2a8994d3c8ecb01b417c57d0d : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"aafe4e5f86424f12bc9528d2d2f6ef3f1851f80b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-10 15:44:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"667955d67b57cf2a19c9e66a02754237d90a2d0b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-10 18:29:56.000000000","message":"Patch Set 4:\n\nLGTM, will +2 in a day or two since there was quite some discussion around this and I like to give everybody the chance to review again.","accounts_in_message":[],"_revision_number":4},{"id":"84dfe28c4a2b562912ca2e6c352212173d16c296","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-10 18:30:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"86141ed933479a4bd39c7535e84e4fa35801c5f5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-03-11 00:52:53.000000000","message":"Patch Set 4:\n\n\u003e LGTM, will +2 in a day or two since there was quite some discussion\n \u003e around this and I like to give everybody the chance to review\n \u003e again.\n\n++\n\nThanks, Andreas.","accounts_in_message":[],"_revision_number":4},{"id":"e875d0be09ec67dd1d211ab87432381e8b127169","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-03-11 15:51:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"aea6ab71131585034bc2d2d66a557a107c857d3f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-12 07:26:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c608dbf633121f97d1e2538ea56157e90c7b75c8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-13 15:34:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1b60f3b8476fd5cde681a065ab3b384cccb9d759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:35:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"85d39be98d8329ab8d8c7898b0c5b2a152c2488b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:44:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ebb04aa338adbcde08ecceeb4a99d044352309dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:44:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/437113dc3cb74f3fbbffbefda6795cf9 : SUCCESS in 7m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0b15160d646f4a048d8d820f2b22252b : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c3d42e32f16bba4109c3310f0a770efaf50742c2","revisions":{"82de81670c9c7484020c997b2d6ab1db4e041c55":{"kind":"REWORK","_number":1,"created":"2020-03-03 21:58:18.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/711114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/711114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/711114/1"}}},"commit":{"parents":[{"commit":"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7","subject":"Restore blank nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-03 21:51:39.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-03 21:51:39.000000000","tz":-360},"subject":"Add a new project and repository for tripleo-ipa","message":"Add a new project and repository for tripleo-ipa\n\nThis commit creates a new project and repository for tripleo-ipa, which\nwill contains ansible roles used to register and deregister OpenStack\nservices from FreeIPA. This is useful for generating certificates and\nusing them for TLS.\n\nChange-Id: I8e0ee3829a9d603c09f146f2a4a9ddc39656b885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82de81670c9c7484020c997b2d6ab1db4e041c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82de81670c9c7484020c997b2d6ab1db4e041c55"}]},"branch":"refs/heads/master"},"79676c472868f7032b57f3009719cf50f4b4707f":{"kind":"REWORK","_number":2,"created":"2020-03-04 15:39:46.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/711114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/711114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/711114/2"}}},"commit":{"parents":[{"commit":"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7","subject":"Restore blank nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-03 21:51:39.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-04 15:37:50.000000000","tz":-360},"subject":"Add a new project and repository for tripleo-ipa","message":"Add a new project and repository for tripleo-ipa\n\nThis commit creates a new project and repository for tripleo-ipa, which\nwill contains ansible roles used to register and deregister OpenStack\nservices from FreeIPA. This is useful for generating certificates and\nusing them for TLS.\n\nChange-Id: I8e0ee3829a9d603c09f146f2a4a9ddc39656b885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/79676c472868f7032b57f3009719cf50f4b4707f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/79676c472868f7032b57f3009719cf50f4b4707f"}]},"branch":"refs/heads/master"},"696da46059bdbfd81225e6c11b07ccdf8a279877":{"kind":"REWORK","_number":3,"created":"2020-03-10 13:54:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/711114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/711114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/711114/3"}}},"commit":{"parents":[{"commit":"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7","subject":"Restore blank nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-03 21:51:39.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-10 13:53:40.000000000","tz":-300},"subject":"Add a new project and repository for tripleo-ipa","message":"Add a new project and repository for tripleo-ipa\n\nThis commit creates a new project and repository for tripleo-ipa, which\nwill contains ansible roles used to register and deregister OpenStack\nservices from FreeIPA. This is useful for generating certificates and\nusing them for TLS.\n\nChange-Id: I8e0ee3829a9d603c09f146f2a4a9ddc39656b885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/696da46059bdbfd81225e6c11b07ccdf8a279877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/696da46059bdbfd81225e6c11b07ccdf8a279877"}]},"branch":"refs/heads/master"},"c3d42e32f16bba4109c3310f0a770efaf50742c2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-03-10 14:25:39.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/711114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/711114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/711114/4"}}},"commit":{"parents":[{"commit":"67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7","subject":"Restore blank nodepool/scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67cbecdfb488bdb0fce82f797abbc3ffdfe75eb7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-03-03 21:51:39.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2020-03-10 14:25:32.000000000","tz":0},"subject":"Add a new project and repository for tripleo-ipa","message":"Add a new project and repository for tripleo-ipa\n\nThis commit creates a new project and repository for tripleo-ipa, which\nwill contain ansible roles used to register and deregister OpenStack\nservices from FreeIPA. This is useful for generating certificates and\nusing them for TLS.\n\nChange-Id: I8e0ee3829a9d603c09f146f2a4a9ddc39656b885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3d42e32f16bba4109c3310f0a770efaf50742c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3d42e32f16bba4109c3310f0a770efaf50742c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
